Skip to content

feat(AvatarStack): add style prop - #5608

Merged
jonrohan merged 4 commits into
mainfrom
hussam-i-am/avatar-stack-style
Jan 28, 2025
Merged

feat(AvatarStack): add style prop#5608
jonrohan merged 4 commits into
mainfrom
hussam-i-am/avatar-stack-style

Conversation

@hussam-i-am

Copy link
Copy Markdown
Contributor

Relates to https://github.com/github/collaboration-workflows-flex/issues/1297

Changelog

New

Changed

  • Noticed that AvatarStack did not have the style prop available, so adding it to be consistent with other components

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@hussam-i-amhussam-i-am self-assigned this Jan 27, 2025
@changeset-bot

changeset-botBot commented Jan 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d273943

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@hussam-i-am
hussam-i-am marked this pull request as ready for review January 27, 2025 20:56
@hussam-i-am
hussam-i-am requested a review from a team as a code ownerJanuary 27, 2025 20:56
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jan 27, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Jan 27, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.45 KB (+0.03% 🔺)
packages/react/dist/browser.umd.js104.83 KB (+0.1% 🔺)

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for doing this! If you have a sec, could you add a test for this, as well?

@jonrohan
jonrohan enabled auto-merge January 27, 2025 23:27
@jonrohan
jonrohan added this pull request to the merge queueJan 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5608 January 27, 2025 23:37 Inactive
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to no response for status checks Jan 27, 2025
@jonrohan
jonrohan enabled auto-merge January 28, 2025 00:40
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5608 January 28, 2025 00:42 Inactive
@jonrohan
jonrohan added this pull request to the merge queueJan 28, 2025
Merged via the queue into main with commit 3a41430Jan 28, 2025
@jonrohan
jonrohan deleted the hussam-i-am/avatar-stack-style branch January 28, 2025 00:57
@primerprimerBot mentioned this pull request Jan 28, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@hussam-i-am@joshblack@jonrohan