Skip to content

bugfix(ActionList.Item): Remove styles from enabled ActionList.Item - #5657

Merged
langermank merged 2 commits into
mainfrom
actionlist-sxprop-leak
Feb 5, 2025
Merged

bugfix(ActionList.Item): Remove styles from enabled ActionList.Item#5657
langermank merged 2 commits into
mainfrom
actionlist-sxprop-leak

Conversation

@langermank

@langermanklangermank commented Feb 4, 2025

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4745
Closeshttps://github.com/github/primer/issues/4752

Changelog

New

Changed

We discovered that the ActionList.Item was applying the old styles from styled components and the new CSS modules classnames whenever a sxProp was being passed in. This was causing some visual bugs in production. Removing the extra style here fixes the issue

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Feb 4, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@langermanklangermank changed the title remove merge stylesRemove styles from enabled ActionList.ItemFeb 4, 2025
@github-actions

github-actionsBot commented Feb 4, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.42 KB (+0.05% 🔺)
packages/react/dist/browser.umd.js105.82 KB (-0.01% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/360875

@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 707fdaa

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@langermank
langermank marked this pull request as ready for review February 4, 2025 22:51
@langermank
langermank requested a review from a team as a code ownerFebruary 4, 2025 22:51

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Just wanted to ask, does style need to be added anywhere else when sx is passed since it's no longer being merged?

@jonrohan

Copy link
Copy Markdown
Member

Just wanted to ask, does style need to be added anywhere else when sx is passed since it's no longer being merged?

Style shouldn't have been included here, it was mistakenly added when the classes takes care of the styles. We found we were applying styled components styles and CSS modules styles in this case

@jonrohanjonrohan changed the title Remove styles from enabled ActionList.Itembugfix(ActionList.Item): Remove styles from enabled ActionList.ItemFeb 4, 2025
@langermank
langermank added this pull request to the merge queueFeb 4, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Feb 4, 2025
@langermank
langermank added this pull request to the merge queueFeb 4, 2025
Merged via the queue into main with commit de4fc48Feb 5, 2025
@langermank
langermank deleted the actionlist-sxprop-leak branch February 5, 2025 00:13
@primerprimerBot mentioned this pull request Feb 5, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@langermank@jonrohan@joshblack