Skip to content

chore(UnderlinePanels): Adding the option to pass in className - #5686

Merged
jonrohan merged 2 commits into
mainfrom
add_classname_prop_underlinepanels
Feb 13, 2025
Merged

chore(UnderlinePanels): Adding the option to pass in className#5686
jonrohan merged 2 commits into
mainfrom
add_classname_prop_underlinepanels

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Changelog

New

The UnderlinePanels component wasn't supporting passing in className. Adding to the prop list

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerFebruary 10, 2025 23:14
@changeset-bot

changeset-botBot commented Feb 10, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b596c3a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Feb 10, 2025
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Feb 10, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions
github-actionsBot requested a deployment to storybook-preview-5686 February 10, 2025 23:17 Abandoned
@github-actions

github-actionsBot commented Feb 10, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.53 KB (0%)
packages/react/dist/browser.umd.js105.9 KB (0%)

@jonrohan
jonrohan added this pull request to the merge queueFeb 11, 2025
@jonrohan
jonrohan removed this pull request from the merge queue due to a manual request Feb 11, 2025
@jonrohan
jonrohan added this pull request to the merge queueFeb 13, 2025
github-merge-queueBot pushed a commit that referenced this pull request Feb 13, 2025
* chore(UnderlinePanels): Adding the option to pass in className
* Create fifty-deers-applaud.md
Merged via the queue into main with commit 64f183bFeb 13, 2025
@jonrohan
jonrohan deleted the add_classname_prop_underlinepanels branch February 13, 2025 17:58
@primerprimerBot mentioned this pull request Feb 13, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
* chore(UnderlinePanels): Adding the option to pass in className
* Create fifty-deers-applaud.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@TylerJDev