Skip to content

Blankslate: Add button support to PrimaryAction - #5687

Merged
TylerJDev merged 7 commits into
mainfrom
tylerjdev/add-button-support-to-blankslate
Feb 13, 2025
Merged

Blankslate: Add button support to PrimaryAction#5687
TylerJDev merged 7 commits into
mainfrom
tylerjdev/add-button-support-to-blankslate

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Feb 11, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4767

Changelog

New

  • Adds button support in Blankslate.PrimaryAction

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Feb 11, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 42e71b9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Feb 11, 2025
@github-actions

github-actionsBot commented Feb 11, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.57 KB (0%)
packages/react/dist/browser.umd.js106.04 KB (0%)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5687 February 11, 2025 21:07 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review February 11, 2025 21:12
@TylerJDev
TylerJDev requested a review from a team as a code ownerFebruary 11, 2025 21:12
@github-actions
github-actionsBot requested a deployment to storybook-preview-5687 February 11, 2025 21:14 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5687 February 11, 2025 21:27 Inactive
Comment threadpackages/react/src/Blankslate/Blankslate.tsx Outdated
Comment threadpackages/react/src/Blankslate/Blankslate.tsx Outdated
TylerJDevand others added 2 commits February 12, 2025 08:47
Co-authored-by: Josh Black <joshblack@github.com>
@github-actions
github-actionsBot requested a deployment to storybook-preview-5687 February 12, 2025 13:51 Abandoned
Comment threadpackages/react/src/Blankslate/Blankslate.tsx Outdated
@github-actions
github-actionsBot requested a deployment to storybook-preview-5687 February 12, 2025 14:02 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5687 February 12, 2025 14:17 Inactive
@TylerJDev
TylerJDev added this pull request to the merge queueFeb 13, 2025
Merged via the queue into main with commit 2bbc5a0Feb 13, 2025
@TylerJDev
TylerJDev deleted the tylerjdev/add-button-support-to-blankslate branch February 13, 2025 17:38
@primerprimerBot mentioned this pull request Feb 13, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
* Add button support to `PrimaryAction` in `BlankSlate`
* Improve story, spread props, fix types
* Add changeset
* Small typo
* Change types
* Update packages/react/src/Blankslate/Blankslate.tsx
Co-authored-by: Josh Black <joshblack@github.com>
* Update packages/react/src/Blankslate/Blankslate.tsx
---------
Co-authored-by: Josh Black <joshblack@github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack