Skip to content

chore: Move components from staff CSS modules feature flag to ga - #5710

Merged
jonrohan merged 19 commits into
mainfrom
move_staff_components_flag_to_ga
Feb 21, 2025
Merged

chore: Move components from staff CSS modules feature flag to ga#5710
jonrohan merged 19 commits into
mainfrom
move_staff_components_flag_to_ga

Conversation

@jonrohan

Copy link
Copy Markdown
Member

Changelog

New

Changed

  • Moving Autocomplete to CSS modules ga flag
  • Moving ConfirmationDialog to primer_react_css_modules_ga feature flag
  • Moving DataTable to primer_react_css_modules_ga feature flag
  • Moving UnderlinePanels to primer_react_css_modules_ga feature flag
  • Moving InlineMessage to primer_react_css_modules_ga feature flag
  • Moving KeybindingHint to primer_react_css_modules_ga feature flag
  • Moving NavList to primer_react_css_modules_ga feature flag
  • Moving Overlay to primer_react_css_modules_ga feature flag
  • Moving PageHeader to primer_react_css_modules_ga feature flag
  • Moving PageLayout to primer_react_css_modules_ga feature flag
  • Moving Pagination to primer_react_css_modules_ga feature flag
  • Moving SegmentedControl to primer_react_css_modules_ga feature flag
  • Moving SelectPanel to primer_react_css_modules_ga feature flag
  • Moving SubNav to primer_react_css_modules_ga feature flag
  • Moving Token to primer_react_css_modules_ga feature flag
  • Moving Tooltip to primer_react_css_modules_ga feature flag
  • Moving BaseStyles to primer_react_css_modules_ga feature flag
  • Moving SideNav to primer_react_css_modules_ga feature flag

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

This shouldn't need any integration testing, prod will remain unchanged, only more eyes on these components.

Merge checklist

@jonrohan
jonrohan requested a review from a team as a code ownerFebruary 19, 2025 22:25
@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 29cddd7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Feb 19, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.77 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js106.17 KB (+0.11% 🔺)

@jonrohan
jonrohan added this pull request to the merge queueFeb 19, 2025
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Feb 19, 2025
@amirdjbdj3873

Copy link
Copy Markdown

@nami1278

@amirdjbdj3873

Copy link
Copy Markdown

کمپپهممثمثمثوثمثم

@joshblack
joshblack added this pull request to the merge queueFeb 21, 2025
@joshblack
joshblack removed this pull request from the merge queue due to a manual request Feb 21, 2025
@jonrohan
jonrohan added this pull request to the merge queueFeb 21, 2025
Merged via the queue into main with commit b1cdc63Feb 21, 2025
@jonrohan
jonrohan deleted the move_staff_components_flag_to_ga branch February 21, 2025 18:47
@primerprimerBot mentioned this pull request Feb 21, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
* Moving Autocomplete to CSS modules ga flag
* Move AvatarStack from staff to ga feature flag
* Moving ConfirmationDialog to `primer_react_css_modules_ga` feature flag
* Moving DataTable to `primer_react_css_modules_ga` feature flag
* Moving UnderlinePanels to `primer_react_css_modules_ga` feature flag
* Moving InlineMessage to `primer_react_css_modules_ga` feature flag
* Moving KeybindingHint to `primer_react_css_modules_ga` feature flag
* Moving NavList to `primer_react_css_modules_ga` feature flag
* Moving Overlay to `primer_react_css_modules_ga` feature flag
* Moving PageHeader to `primer_react_css_modules_ga` feature flag
* Moving PageLayout to `primer_react_css_modules_ga` feature flag
* Moving Pagination to `primer_react_css_modules_ga` feature flag
* Moving SegmentedControl to `primer_react_css_modules_ga` feature flag
* Moving SelectPanel to `primer_react_css_modules_ga` feature flag
* Moving SubNav to `primer_react_css_modules_ga` feature flag
* Moving Token to `primer_react_css_modules_ga` feature flag
* Moving Tooltip to `primer_react_css_modules_ga` feature flag
* Moving BaseStyles to `primer_react_css_modules_ga` feature flag
* Moving SideNav to `primer_react_css_modules_ga` feature flag
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@amirdjbdj3873@joshblack