Skip to content

fix(TextInput): use cursor text on wrapper - #5733

Merged
francinelucca merged 4 commits into
mainfrom
4816-mouse-cursor-not-updated-if-sizelarge
Mar 5, 2025
Merged

fix(TextInput): use cursor text on wrapper#5733
francinelucca merged 4 commits into
mainfrom
4816-mouse-cursor-not-updated-if-sizelarge

Conversation

@francinelucca

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4816

Fixes bug where TextInput size large was displaying cursor type pointer when hovered over the edges because it wasn't hovering the input, but the wrapper span. Adds cursor type text to span.

Changelog

New

  • Add cursor type text to TextInput wrapper span

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@francinelucca
francinelucca requested a review from a team as a code ownerMarch 4, 2025 20:07
@changeset-bot

changeset-botBot commented Mar 4, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fb54035

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Mar 4, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 4, 2025
@github-actions
github-actionsBot requested a deployment to storybook-preview-5733 March 4, 2025 20:10 Abandoned
@github-actions

github-actionsBot commented Mar 4, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.93 KB (-0.03% 🔽)
packages/react/dist/browser.umd.js106.39 KB (+0.07% 🔺)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/365293

@github-actions
github-actionsBot requested a deployment to storybook-preview-5733 March 4, 2025 20:20 Abandoned
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 4, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5733 March 4, 2025 20:33 Inactive
@francinelucca
francinelucca added this pull request to the merge queueMar 5, 2025
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Mar 5, 2025
@francinelucca
francinelucca added this pull request to the merge queueMar 5, 2025
Merged via the queue into main with commit 7849c95Mar 5, 2025
@francinelucca
francinelucca deleted the 4816-mouse-cursor-not-updated-if-sizelarge branch March 5, 2025 00:47
@primerprimerBot mentioned this pull request Mar 4, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff memberstatus: review needed

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@joshblack