Skip to content

Move ActionList feature flag to GA - #5743

Merged
langermank merged 2 commits into
mainfrom
al-ga
Mar 7, 2025
Merged

Move ActionList feature flag to GA#5743
langermank merged 2 commits into
mainfrom
al-ga

Conversation

@langermank

Copy link
Copy Markdown
Contributor

Move ActionList FF to GA

@langermank
langermank requested a review from a team as a code ownerMarch 7, 2025 15:22
@changeset-bot

changeset-botBot commented Mar 7, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a18f9bc

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 7, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@langermank
langermank enabled auto-merge March 7, 2025 15:24
@github-actions
github-actionsBot requested a deployment to storybook-preview-5743 March 7, 2025 15:25 Abandoned
@github-actions

github-actionsBot commented Mar 7, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js106.02 KB (+0.11% 🔺)
packages/react/dist/browser.umd.js106.42 KB (+0.08% 🔺)

@langermank
langermank added this pull request to the merge queueMar 7, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5743 March 7, 2025 15:39 Inactive
Merged via the queue into main with commit 9d78410Mar 7, 2025
@langermank
langermank deleted the al-ga branch March 7, 2025 15:48
@primerprimerBot mentioned this pull request Mar 7, 2025
hectahertz pushed a commit that referenced this pull request Mar 20, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@francinelucca