Skip to content

Add -- prefix to CSS Custom Property for SkeletonBox - #5768

Merged
joshblack merged 2 commits into
mainfrom
joshblack-patch-4
Mar 13, 2025
Merged

Add -- prefix to CSS Custom Property for SkeletonBox#5768
joshblack merged 2 commits into
mainfrom
joshblack-patch-4

Conversation

@joshblack

Copy link
Copy Markdown
Member

Update the CSS Custom Property to be properly formatted.

Changelog

New

Changed

  • Update base-duration to --base-duration

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

@joshblack
joshblack requested a review from a team as a code ownerMarch 13, 2025 14:58
@joshblack
joshblack requested a review from jonrohanMarch 13, 2025 14:58
@changeset-bot

changeset-botBot commented Mar 13, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1e0a65a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 13, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 13, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js106.05 KB (0%)
packages/react/dist/browser.umd.js106.33 KB (0%)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5768 March 13, 2025 15:04 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5768 March 13, 2025 15:19 Inactive
@joshblack
joshblack added this pull request to the merge queueMar 13, 2025
Merged via the queue into main with commit 7b2eb4eMar 13, 2025
@joshblack
joshblack deleted the joshblack-patch-4 branch March 13, 2025 18:20
@primerprimerBot mentioned this pull request Mar 13, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@francinelucca