Skip to content

Remove the CSS module feature flag from SkeletonAvatar SkeletonBox and SkeletonText - #5787

Merged
jonrohan merged 2 commits into
mainfrom
css_modules_remove_flag_skeleton
Mar 20, 2025
Merged

Remove the CSS module feature flag from SkeletonAvatar SkeletonBox and SkeletonText#5787
jonrohan merged 2 commits into
mainfrom
css_modules_remove_flag_skeleton

Conversation

@jonrohan

@jonrohanjonrohan commented Mar 17, 2025

Copy link
Copy Markdown
Member

This PR removes the CSS modules feature flag from the Skeleton component. The component Skeleton and sub components are used 19 times in dotcom.

Changelog

New

Changed

Removed

Removes the CSS modules feature flag from the Skeleton components.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Using Integration testing.

Merge checklist

@changeset-bot

changeset-botBot commented Mar 17, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 842d582

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 17, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 17, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.39 KB (-0.49% 🔽)
packages/react/dist/browser.umd.js105.61 KB (-0.5% 🔽)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/367616

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5787 March 18, 2025 00:03 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 18, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5787 March 18, 2025 00:16 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5787 March 18, 2025 18:53 Inactive
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Mar 18, 2025
@jonrohan

Copy link
Copy Markdown
MemberAuthor

@jonrohan
jonrohan marked this pull request as ready for review March 18, 2025 19:55
@jonrohan
jonrohan requested a review from a team as a code ownerMarch 18, 2025 19:55
@jonrohan
jonrohan requested a review from TylerJDevMarch 18, 2025 19:55
@jonrohan
jonrohan added this pull request to the merge queueMar 20, 2025
Merged via the queue into main with commit 9272a8cMar 20, 2025
@jonrohan
jonrohan deleted the css_modules_remove_flag_skeleton branch March 20, 2025 21:37
@primerprimerBot mentioned this pull request Mar 20, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@TylerJDev