Skip to content

chore(Table): Remove the CSS modules feature flag from the Table component - #5788

Merged
jonrohan merged 7 commits into
mainfrom
remove_css_modules_flag_table
Apr 11, 2025
Merged

chore(Table): Remove the CSS modules feature flag from the Table component#5788
jonrohan merged 7 commits into
mainfrom
remove_css_modules_flag_table

Conversation

@jonrohan

@jonrohanjonrohan commented Mar 18, 2025

Copy link
Copy Markdown
Member

This PR removes the CSS modules feature flag from the Table component. The component Table and sub components are used 121 times in dotcom.

Closeshttps://github.com/github/primer/issues/4307

Changelog

New

Changed

Removed

Removes the CSS modules feature flag from the Table components.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Using Integration testing.

Merge checklist

@changeset-bot

changeset-botBot commented Mar 18, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 83fa205

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 18, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 18, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.48 KB (0%)
packages/react/dist/browser.umd.js104.86 KB (0%)

Comment threadpackages/react/src/DataTable/Table.tsx Outdated
@jonrohanjonrohan changed the title Remove the CSS modules feature flag from Tablechore(Table): Remove the CSS modules feature flag from the Table componentApr 4, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5788 April 4, 2025 23:59 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5788 April 5, 2025 00:13 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 5, 2025
@github-actions
github-actionsBot requested a deployment to storybook-preview-5788 April 10, 2025 17:47 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5788 April 10, 2025 18:05 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/371527

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Apr 10, 2025
@jonrohan
jonrohan marked this pull request as ready for review April 10, 2025 21:22
@jonrohan
jonrohan requested a review from a team as a code ownerApril 10, 2025 21:22
@jonrohan
jonrohan added this pull request to the merge queueApr 11, 2025
Merged via the queue into main with commit cd9987dApr 11, 2025
@jonrohan
jonrohan deleted the remove_css_modules_flag_table branch April 11, 2025 16:39
@primerprimerBot mentioned this pull request Apr 11, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@joshblack@francinelucca