Skip to content

Remove CSS modules feature flag from the InlineMessage component - #5804

Merged
jonrohan merged 1 commit into
mainfrom
css_modules_remove_flag_inlinemessage
Mar 25, 2025
Merged

Remove CSS modules feature flag from the InlineMessage component#5804
jonrohan merged 1 commit into
mainfrom
css_modules_remove_flag_inlinemessage

Conversation

@jonrohan

@jonrohanjonrohan commented Mar 21, 2025

Copy link
Copy Markdown
Member

This PR removes the CSS modules feature flag from the InlineMessage component. The component InlineMessage is used 5 times in dotcom.

Changelog

New

Changed

Removed

Removes the CSS modules feature flag from the InlineMessage components.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Using Integration testing.

Merge checklist

@changeset-bot

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: e2c5f37

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 21, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.38 KB (0%)
packages/react/dist/browser.umd.js105.74 KB (0%)

@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/368656

@primer-integration

Copy link
Copy Markdown

🔴 golden-jobs completed with status failure.

@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 21, 2025
@jonrohan

Copy link
Copy Markdown
MemberAuthor

The failures are related to a failure that's currently in main

@jonrohan
jonrohan marked this pull request as ready for review March 24, 2025 22:23
@jonrohan
jonrohan requested a review from a team as a code ownerMarch 24, 2025 22:23
@jonrohanjonrohan added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Mar 24, 2025

@francineluccafrancinelucca left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:shipit:

@jonrohan
jonrohan added this pull request to the merge queueMar 25, 2025
Merged via the queue into main with commit 687343cMar 25, 2025
@jonrohan
jonrohan deleted the css_modules_remove_flag_inlinemessage branch March 25, 2025 19:19
@primerprimerBot mentioned this pull request Mar 25, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@francinelucca