Skip to content

Adds util function to account for SX dependent components - #5821

Merged
jonrohan merged 3 commits into
mainfrom
jshark/toggle-sx-component
Mar 25, 2025
Merged

Adds util function to account for SX dependent components#5821
jonrohan merged 3 commits into
mainfrom
jshark/toggle-sx-component

Conversation

@jamieshark

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/4945

Changelog

New

  • Adds a new util function toggleSxComponent as well as accompanying tests.

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 25, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1050f3b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 25, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@jamiesharkjamieshark self-assigned this Mar 25, 2025
@jamieshark
jamieshark marked this pull request as ready for review March 25, 2025 21:10
@jamieshark
jamieshark requested a review from a team as a code ownerMarch 25, 2025 21:10
@github-actions

github-actionsBot commented Mar 25, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.78 KB (0%)
packages/react/dist/browser.umd.js106.13 KB (0%)

@jonrohanjonrohan left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Nice love it 👍🏻

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5821 March 25, 2025 22:05 Inactive
@jonrohan
jonrohan added this pull request to the merge queueMar 25, 2025
Merged via the queue into main with commit 3a102a7Mar 25, 2025
@jonrohan
jonrohan deleted the jshark/toggle-sx-component branch March 25, 2025 23:17
@primerprimerBot mentioned this pull request Mar 25, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jamieshark@jonrohan