Skip to content

feat(UnderlineNav): Support inset and flush variants - #5829

Merged
maraisr merged 10 commits into
mainfrom
mr/padded-tabs
Jun 16, 2025
Merged

feat(UnderlineNav): Support inset and flush variants#5829
maraisr merged 10 commits into
mainfrom
mr/padded-tabs

Conversation

@maraisr

@maraisrmaraisr commented Mar 27, 2025

Copy link
Copy Markdown
Contributor

Variant: flush

image

Variant: inset (default)

image

Changelog

New

Added inset and full as variant props to UnderlineNav to aid in removing horizontal spacing.

Rollout strategy

  • Patch release

Testing & Reviewing

Since the current behaviour is with inset spacing, the default for this new prop is also inset. So fully backwards compatible, a testing strategy is that all existing vrt tests pass.

Merge checklist

@changeset-bot

changeset-botBot commented Mar 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f93b66b

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 27, 2025
@github-actions

github-actionsBot commented Mar 27, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js94.56 KB (+0.06% 🔺)
packages/react/dist/browser.umd.js94.64 KB (-0.01% 🔽)

Comment threadpackages/react/src/experimental/UnderlinePanels/UnderlinePanels.tsx Outdated

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Heads up that we're about to remove the CSS Modules feature flag from UnderlinePanel early next week. I would hold off on this until that merges so that you don't need to extend this behavior to sx

@lesliecdubs

lesliecdubs commented Apr 26, 2025

Copy link
Copy Markdown
Member

Hey @maraisr, thanks for the PR! Is this ready for review once conflicts are resolved?

@maraisr

Copy link
Copy Markdown
ContributorAuthor

Hi @lesliecdubs, I don't think its ready yet. Failing tests, conflicts etc. I have not had much time to get this PR up to shape, so if somebody on the team is looking to take this on, please do! 🚀 If not, I'll dedicate some time by the end of this week to wrap this up. 👍 I think its a great feature, and unlocks the removal of some sx usages in product code!

@hectahertz
hectahertz self-requested a review May 19, 2025 14:25
/* stylelint-disable-next-line primer/box-shadow */
box-shadow: inset 0 -1px var(--borderColor-muted);

&:where([data-flush="false"], &[data-flush="false"] &) {

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

If you could add a small explanation of the selector that would be nice, not required

Comment threadpackages/react/src/UnderlineNav/UnderlineNav.tsx Outdated
@hectahertz

Copy link
Copy Markdown
Contributor

@maraisr if you have some time to make those changes I'd love to get this merged!

@github-actions
github-actionsBot requested a deployment to storybook-preview-5829 May 29, 2025 00:00 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5829 May 29, 2025 00:13 Inactive
@maraisrmaraisr changed the title feat: support the ability to remove inline paddingfeat(UnderlineNav): Support inset and full variantsMay 29, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5829 May 29, 2025 01:58 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels May 29, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5829 May 29, 2025 06:01 Inactive

@hectahertzhectahertz left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

This looks great! Thanks for your contribution!! 🚀

I think flush would work better, as that's what we've used in other components. If you could please change that, then we can merge this :)

Approving in advance!

)
}

export const Full = () => {

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Suggested change
exportconstFull=()=>{
exportconstVariantFull=()=>{

controls: {
expanded: true,
// variant and size are developed in the first design iteration but then they are abondened.
// size and others were developed in the first design iteration but then they are abandoned.

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

abondened

lol

@maraisrmaraisr changed the title feat(UnderlineNav): Support inset and full variantsfeat(UnderlineNav): Support inset and flush variantsMay 29, 2025
@maraisrmaraisr added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label May 29, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label May 29, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5829 May 29, 2025 10:09 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5829 May 29, 2025 10:23 Inactive
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/380428

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Jun 6, 2025
@maraisr
maraisr added this pull request to the merge queueJun 16, 2025
Merged via the queue into main with commit 72ae1e2Jun 16, 2025
@maraisr
maraisr deleted the mr/padded-tabs branch June 16, 2025 02:08
@primerprimerBot mentioned this pull request Jun 13, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@maraisr@lesliecdubs@hectahertz@langermank