Skip to content

chore(SelectPanel): put fullscreen functionality behind FF - #5832

Merged
TylerJDev merged 6 commits into
mainfrom
fix/selectpanel-add-feature-flag
Mar 27, 2025
Merged

chore(SelectPanel): put fullscreen functionality behind FF#5832
TylerJDev merged 6 commits into
mainfrom
fix/selectpanel-add-feature-flag

Conversation

@francinelucca

@francineluccafrancinelucca commented Mar 27, 2025

Copy link
Copy Markdown
Member

Relates to https://github.com/github/primer/issues/4109

Changelog

Puts SelectPanel's fullscreen, save, cancel and close button functionality behind new Feature Flag

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@francinelucca
francinelucca requested a review from a team as a code ownerMarch 27, 2025 21:58
@changeset-bot

changeset-botBot commented Mar 27, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 238e7bd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 27, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Mar 27, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.64 KB (+0.08% 🔺)
packages/react/dist/browser.umd.js106.06 KB (+0.13% 🔺)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5832 March 27, 2025 22:03 Abandoned
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5832 March 27, 2025 22:15 Inactive
@primer
primerBot requested a review from a team as a code ownerMarch 27, 2025 22:22
@primer
primerBot requested a review from mperrottiMarch 27, 2025 22:22
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5832 March 27, 2025 22:35 Inactive
@francineluccafrancinelucca added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Mar 27, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5832 March 27, 2025 22:49 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5832 March 27, 2025 23:03 Inactive
@TylerJDev
TylerJDev merged commit b6c7690 into mainMar 27, 2025
@TylerJDev
TylerJDev deleted the fix/selectpanel-add-feature-flag branch March 27, 2025 23:05
@primerprimerBot mentioned this pull request Mar 26, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@francinelucca@TylerJDev