Skip to content

Remove the CSS modules feature flag from the FormControl component - #5851

Merged
jamieshark merged 18 commits into
mainfrom
css_modules_remove_flag_formcontrol
Apr 2, 2025
Merged

Remove the CSS modules feature flag from the FormControl component#5851
jamieshark merged 18 commits into
mainfrom
css_modules_remove_flag_formcontrol

Conversation

@jamieshark

Copy link
Copy Markdown
Contributor

This PR removes the CSS modules feature flag from the FormControl component. The component FormControl is used 509 times in dotcom.

Changelog

New

Changed

Removed

Removes the CSS modules feature flag from the FormControl component.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 31, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f53fd68

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Mar 31, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@jamieshark
jamieshark marked this pull request as ready for review March 31, 2025 16:18
@jamieshark
jamieshark requested a review from a team as a code ownerMarch 31, 2025 16:18
@github-actions

github-actionsBot commented Mar 31, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.99 KB (-0.52% 🔽)
packages/react/dist/browser.umd.js105.38 KB (-0.42% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5851 March 31, 2025 16:30 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 31, 2025
Comment on lines +34 to +35
<Text
className={clsx({[classes.InputValidation]: sx !== defaultSxProp})}

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

confused by this line. these styles are only needed if sx prop is supplied?

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

ah I think I was confused here, thinking that we would only apply the modules styles if we didn't have custom sx, but I think the idea is that we always apply the styles anyway. I'll make this change!

Comment threadpackages/react/src/FormControl/__tests__/FormControl.test.tsx Outdated
@github-actions
github-actionsBot requested a deployment to storybook-preview-5851 April 2, 2025 17:15 Abandoned
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/370943

@primer
primerBot requested a review from a team as a code ownerApril 2, 2025 19:03
@primer
primerBot requested a review from langermankApril 2, 2025 19:03
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5851 April 2, 2025 19:17 Inactive
@jamiesharkjamieshark added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Apr 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5851 April 2, 2025 20:03 Inactive
@jamiesharkjamieshark added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 2, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5851 April 2, 2025 21:06 Inactive
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Apr 2, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5851 April 2, 2025 21:21 Inactive
@jamieshark
jamieshark added this pull request to the merge queueApr 2, 2025
Merged via the queue into main with commit a812a86Apr 2, 2025
@jamieshark
jamieshark deleted the css_modules_remove_flag_formcontrol branch April 2, 2025 22:17
@primerprimerBot mentioned this pull request Apr 2, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jamieshark@francinelucca