Skip to content

bugfix(PageHeader): Use font-weight initial to match the styled component styling - #5852

Merged
jonrohan merged 4 commits into
mainfrom
pageheader_description_initial
Mar 31, 2025
Merged

bugfix(PageHeader): Use font-weight initial to match the styled component styling #5852
jonrohan merged 4 commits into
mainfrom
pageheader_description_initial

Conversation

@jonrohan

@jonrohanjonrohan commented Mar 31, 2025

Copy link
Copy Markdown
Member

It was brought up in https://github.com/orgs/community/discussions/154025 that our commit headers were looking thin. Upon investigation I found that the old style had font-weight: initial;

And the new style had a primitives variable.

font-weight:var(--base-text-weight-light);

Changelog

New

Changed

Use font-weight: initial in PageHeader description to match the styled component styling.

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 31, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: fd6cfbf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Mar 31, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@jonrohan
jonrohan marked this pull request as ready for review March 31, 2025 16:42
@jonrohan
jonrohan requested a review from a team as a code ownerMarch 31, 2025 16:42
@github-actions

github-actionsBot commented Mar 31, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js105.92 KB (0%)
packages/react/dist/browser.umd.js106.2 KB (0%)

@jonrohan
jonrohan enabled auto-merge March 31, 2025 16:43
@github-actions
github-actionsBot requested a deployment to storybook-preview-5852 March 31, 2025 16:45 Abandoned
@joshblack

Copy link
Copy Markdown
Member

@jonrohan is there a good VRT test we could add for this or is it already included?

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5852 March 31, 2025 17:05 Inactive
@jonrohan

Copy link
Copy Markdown
MemberAuthor

@jonrohan is there a good VRT test we could add for this or is it already included?

Included with this

expect(awaitpage.screenshot()).toMatchSnapshot(`PageHeader.With Description Slot.${theme}.png`)
Surprised it wasn't caught, maybe cause the change is too subtle?

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5852 March 31, 2025 17:47 Inactive
@jonrohan
jonrohan added this pull request to the merge queueMar 31, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5852 March 31, 2025 18:02 Inactive
auto-merge was automatically disabled March 31, 2025 18:06

Pull Request is not mergeable

@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Mar 31, 2025
@jonrohan
jonrohan added this pull request to the merge queueMar 31, 2025
Merged via the queue into main with commit 5d66a4cMar 31, 2025
@jonrohan
jonrohan deleted the pageheader_description_initial branch March 31, 2025 18:25
@primerprimerBot mentioned this pull request Mar 31, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@jonrohan@joshblack@francinelucca