Skip to content

chore(Breadcrumbs): Remove the CSS modules feature flag from the Breadcrumbs component - #5874

Merged
jonrohan merged 4 commits into
mainfrom
remove_css_modules_flag_breadcrumbs
Apr 14, 2025
Merged

chore(Breadcrumbs): Remove the CSS modules feature flag from the Breadcrumbs component#5874
jonrohan merged 4 commits into
mainfrom
remove_css_modules_flag_breadcrumbs

Conversation

@jonrohan

@jonrohanjonrohan commented Apr 3, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/4032

Changelog

New

Changed

Removed

Remove the CSS modules feature flag from the Breadcrumbs component

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Apr 3, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: c9bfed4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Apr 3, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Apr 3, 2025
@github-actions

github-actionsBot commented Apr 3, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.2 KB (-0.27% 🔽)
packages/react/dist/browser.umd.js104.56 KB (-0.29% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5874 April 3, 2025 21:28 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 3, 2025
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/371313

@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Apr 10, 2025
@jonrohan
jonrohan marked this pull request as ready for review April 10, 2025 20:42
@jonrohan
jonrohan requested a review from a team as a code ownerApril 10, 2025 20:42
@jonrohan
jonrohan requested a review from joshblackApril 10, 2025 20:42
@jonrohan
jonrohan enabled auto-merge April 11, 2025 18:17
@jonrohan
jonrohan added this pull request to the merge queueApr 14, 2025
Merged via the queue into main with commit 365d53dApr 14, 2025
@jonrohan
jonrohan deleted the remove_css_modules_flag_breadcrumbs branch April 14, 2025 16:03
@primerprimerBot mentioned this pull request Apr 14, 2025
hectahertz pushed a commit that referenced this pull request May 20, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@jonrohan@langermank