Skip to content

fix(SelectPanel): always show x button on single select - #5912

Merged
TylerJDev merged 2 commits into
mainfrom
fix/close-on-single-select
Apr 10, 2025
Merged

fix(SelectPanel): always show x button on single select#5912
TylerJDev merged 2 commits into
mainfrom
fix/close-on-single-select

Conversation

@francinelucca

Copy link
Copy Markdown
Member

Small improvement to onCancel behavior. Always showing the close icon on single select (otherwise the user gets "stuck" and the only way to dismiss the SelectPanel is by making a selection.

Changelog

Changed

  • Always show x button on single select

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@francinelucca
francinelucca requested a review from a team as a code ownerApril 10, 2025 14:34
@changeset-bot

changeset-botBot commented Apr 10, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: beb39ba

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 10, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Apr 10, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js104.48 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js104.86 KB (+0.07% 🔺)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5912 April 10, 2025 14:41 Abandoned
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/372429

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5912 April 10, 2025 14:55 Inactive
@github-actionsgithub-actionsBot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 10, 2025
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: failing Changes in this PR cause breaking changes in gh/gh labels Apr 10, 2025
@TylerJDev
TylerJDev added this pull request to the merge queueApr 10, 2025
Merged via the queue into main with commit 81821dbApr 10, 2025
@TylerJDev
TylerJDev deleted the fix/close-on-single-select branch April 10, 2025 17:16
@primerprimerBot mentioned this pull request Apr 10, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@francinelucca@hectahertz@TylerJDev