Skip to content

fix(SelectPanel): use dvh units for full screen when available - #5930

Merged
francinelucca merged 2 commits into
mainfrom
fix/selectpanel-mobile-full-height
Apr 16, 2025
Merged

fix(SelectPanel): use dvh units for full screen when available#5930
francinelucca merged 2 commits into
mainfrom
fix/selectpanel-mobile-full-height

Conversation

@francinelucca

@francineluccafrancinelucca commented Apr 16, 2025

Copy link
Copy Markdown
Member

Fixes bug on iPhone were 100vh doesn't make all the content visible since it is rendering under the browser toolbar.

Changelog

New

  • CSS styles for using dvh units when supported in full screen

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Apr 16, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6d33982

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Apr 16, 2025
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Apr 16, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Apr 16, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js102.66 KB (0%)
packages/react/dist/browser.umd.js102.96 KB (0%)

@github-actions
github-actionsBot requested a deployment to storybook-preview-5930 April 16, 2025 01:43 Abandoned
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github! Your integration PR is ready: https://github.com/github/github/pull/373340

@github-actions
github-actionsBottemporarily deployed to storybook-preview-5930 April 16, 2025 01:57 Inactive
@francineluccafrancinelucca added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 16, 2025
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Apr 16, 2025
@primer-integration

Copy link
Copy Markdown

🟢 golden-jobs completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Apr 16, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-5930 April 16, 2025 02:22 Inactive
@francinelucca
francinelucca marked this pull request as ready for review April 16, 2025 02:38
@francinelucca
francinelucca requested a review from a team as a code ownerApril 16, 2025 02:38
@francinelucca
francinelucca added this pull request to the merge queueApr 16, 2025
Merged via the queue into main with commit 876b822Apr 16, 2025
@francinelucca
francinelucca deleted the fix/selectpanel-mobile-full-height branch April 16, 2025 18:05
@primerprimerBot mentioned this pull request Apr 16, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@francinelucca@joshblack@TylerJDev