Skip to content

ActionList: Add new prop disableFocusZone - #6116

Merged
TylerJDev merged 2 commits into
mainfrom
tylerjdev/action-list-disable-focus-zone
Jun 12, 2025
Merged

ActionList: Add new prop disableFocusZone#6116
TylerJDev merged 2 commits into
mainfrom
tylerjdev/action-list-disable-focus-zone

Conversation

@TylerJDev

@TylerJDevTylerJDev commented May 23, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/5227

Changelog

New

  • Adds new prop disableFocusZone which allows consumers to disable the focusZone that is applied to the ActionList component

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented May 23, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5f38528

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels May 23, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented May 23, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js94.57 KB (+0.07% 🔺)
packages/react/dist/browser.umd.js94.7 KB (+0.07% 🔺)

showDividers = false,
role,
sx: sxProp = defaultSxProp,
disableFocusZone = false,

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

We could instead add a prop named focusZoneSettings, which allows consumers to customize the focus zone. This would work much like how it does in AnchoredOverlay.

This might not be needed now, but gives consumers more flexibility with the component, and removes the need for the disableFocusZone boolean prop.

@TylerJDev
TylerJDev marked this pull request as ready for review June 2, 2025 17:53
@TylerJDev
TylerJDev requested a review from a team as a code ownerJune 2, 2025 17:53
@TylerJDev
TylerJDev requested a review from hectahertzJune 2, 2025 17:53
@TylerJDev
TylerJDev added this pull request to the merge queueJun 12, 2025
Merged via the queue into main with commit 4cca0e9Jun 12, 2025
@TylerJDev
TylerJDev deleted the tylerjdev/action-list-disable-focus-zone branch June 12, 2025 11:39
@primerprimerBot mentioned this pull request Jun 12, 2025
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@jonrohan