Skip to content

Remove support for sx and Box usage from the Text component - #6729

Merged
francinelucca merged 28 commits into
mainfrom
liuliu/remove-sx-from-text
Oct 1, 2025
Merged

Remove support for sx and Box usage from the Text component#6729
francinelucca merged 28 commits into
mainfrom
liuliu/remove-sx-from-text

Conversation

@liuliu-dev

@liuliu-dev liuliu-dev commented Sep 4, 2025

Copy link
Copy Markdown
Contributor

Wait until this PR is merged in github-ui: https://github.com/github/github-ui/pull/3366

Closes #5830

sx usage

Changelog

Removed

Removed sx props and Box from Text.

Rollout strategy

  • Major release; if selected, include a written rollout or migration plan

@changeset-bot

changeset-bot Bot commented Sep 4, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 16a5344

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
Name Type
@primer/react Major
@primer/styled-react Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions github-actions Bot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 4, 2025
@github-actions

github-actions Bot commented Sep 4, 2025

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions
github-actions Bot temporarily deployed to storybook-preview-6729 September 4, 2025 23:46 Inactive
@liuliu-dev liuliu-dev closed this Sep 10, 2025
@liuliu-dev
liuliu-dev deleted the liuliu/remove-sx-from-text branch September 10, 2025 20:21
@liuliu-dev
liuliu-dev restored the liuliu/remove-sx-from-text branch September 19, 2025 17:08
@liuliu-dev liuliu-dev reopened this Sep 19, 2025
@github-actions
github-actions Bot requested a deployment to storybook-preview-6729 September 19, 2025 20:20 Abandoned
@github-actions
github-actions Bot temporarily deployed to storybook-preview-6729 September 19, 2025 20:31 Inactive
@github-actions
github-actions Bot temporarily deployed to storybook-preview-6729 September 30, 2025 21:46 Inactive
@github-actions github-actions Bot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 30, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions github-actions Bot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 30, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actions Bot requested a deployment to storybook-preview-6729 September 30, 2025 22:09 Abandoned
@github-actions github-actions Bot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 30, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actions Bot requested a deployment to storybook-preview-6729 September 30, 2025 22:16 Abandoned
@github-actions github-actions Bot removed the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 30, 2025
@github-actions
github-actions Bot temporarily deployed to storybook-preview-6729 September 30, 2025 22:24 Inactive
@github-actions github-actions Bot added integration-tests: failing Changes in this PR cause breaking changes in gh/gh and removed integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh labels Sep 30, 2025
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github-ui! Your integration PR is ready: https://github.com/github/github-ui/pull/3600

@primer-integration

Copy link
Copy Markdown

🟢 ci completed with status success.

@primer primer Bot mentioned this pull request Oct 1, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants