Skip to content

Popover: Remove sx from component - #6754

Merged
TylerJDev merged 13 commits into
mainfrom
tylerjdev/remove-sx-form-popover
Sep 12, 2025
Merged

Popover: Remove sx from component#6754
TylerJDev merged 13 commits into
mainfrom
tylerjdev/remove-sx-form-popover

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Sep 8, 2025

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/5838

Removes sx support in Popover

Changelog

Removed

  • Remove sx references in Popover

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Sep 8, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9d888a6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 2 packages
NameType
@primer/reactMajor
@primer/styled-reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added staff Author is a staff member integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 8, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, this pull request contains changes to the source code that github/github depends on. If you are GitHub staff, we recommend testing these changes with github/github using the integration workflow. Thanks!

@github-actions

github-actionsBot commented Sep 8, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js89.52 KB (+0.04% 🔺)
packages/react/dist/browser.umd.js89.62 KB (+0.03% 🔺)

@github-actions
github-actionsBot requested a deployment to storybook-preview-6754 September 8, 2025 14:52 Abandoned
@primer-integration

Copy link
Copy Markdown

👋 Hi from github/github-ui! Your integration PR is ready: https://github.com/github/github-ui/pull/1960

@primer-integration

Copy link
Copy Markdown

🟢 ci completed with status success.

@github-actionsgithub-actionsBot added integration-tests: passing Changes in this PR do NOT cause breaking changes in gh/gh and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 8, 2025
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 8, 2025
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Sep 8, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@TylerJDevTylerJDev added update snapshots 🤖 Command that updates VRT snapshots on the pull request and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request labels Sep 8, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6754 September 8, 2025 15:38 Inactive
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 8, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actions
github-actionsBottemporarily deployed to storybook-preview-6754 September 12, 2025 13:06 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6754 September 12, 2025 13:16 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 12, 2025
@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed update snapshots 🤖 Command that updates VRT snapshots on the pull request integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 12, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 12, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@github-actionsgithub-actionsBot added integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm and removed integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm labels Sep 12, 2025
@github-actions

Copy link
Copy Markdown
Contributor

👋 Hi, there are new commits since the last successful integration test. We recommend running the integration workflow once more, unless you are sure the new changes do not affect github/github. Thanks!

@TylerJDev
TylerJDev added this pull request to the merge queueSep 12, 2025
@github-actions
github-actionsBottemporarily deployed to storybook-preview-6754 September 12, 2025 16:40 Inactive
Merged via the queue into main with commit 4d080aaSep 12, 2025
43 checks passed
@TylerJDev
TylerJDev deleted the tylerjdev/remove-sx-form-popover branch September 12, 2025 16:46
@primerprimerBot mentioned this pull request Sep 12, 2025
pksjce pushed a commit that referenced this pull request Sep 18, 2025
Co-authored-by: TylerJDev <26746305+TylerJDev@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

integration-tests: passingChanges in this PR do NOT cause breaking changes in gh/ghintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpmstaffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@TylerJDev@llastflowers