Skip to content

Fix UnderlineNav not re-rendering when child tabs change - #8037

Closed
liuliu-dev wants to merge 8 commits into
mainfrom
liuliu/fix-underlinenav-re-rendering
Closed

Fix UnderlineNav not re-rendering when child tabs change#8037
liuliu-dev wants to merge 8 commits into
mainfrom
liuliu/fix-underlinenav-re-rendering

Conversation

@liuliu-dev

Copy link
Copy Markdown
Contributor

Closeshttps://github.com/github/primer/issues/6389

Changelog

New

Changed

Removed

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Jun 23, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ce751db

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@liuliu-devliuliu-dev added the Canary Release Apply this label when you want CI to create a canary release of the current PR label Jun 23, 2026
@github-actionsgithub-actionsBot added the integration-tests: recommended This change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm label Jun 23, 2026
@github-actions

Copy link
Copy Markdown
Contributor

⚠️ Action required

👋 Hi, this pull request contains changes to the source code that github/github-ui depends on. If you are GitHub staff, test these changes with github/github-ui using the integration workflow. If this doesn't work, you can also use the original workflow here. Check the integration testing docs for step-by-step instructions. Or, apply the integration-tests: skipped manually label to skip these checks.

To publish a canary release for integration testing, apply the Canary Release label to this PR.

@github-actions
github-actionsBot requested a deployment to storybook-preview-8037 June 23, 2026 20:46 Abandoned
@github-actions
github-actionsBottemporarily deployed to storybook-preview-8037 June 23, 2026 21:11 Inactive
@primer-integration

primer-integrationBot commented Jun 24, 2026

Copy link
Copy Markdown

Integration test results from github/github-ui PR:

Passed CI  Passed
Waiting VRT   Waiting
Passed Projects  Passed

@liuliu-dev

Copy link
Copy Markdown
ContributorAuthor

Closing in favor of #7506

@primer
primerBot deleted the liuliu/fix-underlinenav-re-rendering branch June 30, 2026 18:09
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Canary ReleaseApply this label when you want CI to create a canary release of the current PRintegration-tests: recommendedThis change needs to be tested for breaking changes. See https://arc.net/l/quote/tdmpakpm

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@liuliu-dev