Skip to content
This repository was archived by the owner on Dec 15, 2021. It is now read-only.

Don't move 'README.md' over 'index.md', either - #161

Merged
knative-prow-robot merged 1 commit into
knative:masterfrom
evankanderson:in-dex
Apr 15, 2020
Merged

Don't move 'README.md' over 'index.md', either#161
knative-prow-robot merged 1 commit into
knative:masterfrom
evankanderson:in-dex

Conversation

@evankanderson

Copy link
Copy Markdown
Member

This is needed for knative/docs#2377 to render correctly -- otherwise it overwrites the index.md files with the README.md files without appropriate frontmatter.

@googlebotgooglebot added the cla: yes Indicates the PR's author has signed the CLA. label Apr 15, 2020
@knative-prow-robotknative-prow-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Apr 15, 2020
@evankanderson

Copy link
Copy Markdown
MemberAuthor

/assign @n3wscott , since we don't really have a maintainer here. :-/

@n3wscottn3wscott left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@knative-prow-robotknative-prow-robot added the lgtm Indicates that a PR is ready to be merged. label Apr 15, 2020
@knative-prow-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: evankanderson, n3wscott

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

Sign up for freeto subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approvedIndicates a PR has been approved by an approver from all required OWNERS files.cla: yesIndicates the PR's author has signed the CLA.lgtmIndicates that a PR is ready to be merged.size/XSDenotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants

@evankanderson@knative-prow-robot@n3wscott@googlebot