Uh oh!
There was an error while loading. Please reload this page.
Conversation
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
Codecov Report❌ Patch coverage is 📢 Thoughts on this report? Let us know! |
Lunaria Status Overview🌕 This pull request will trigger status changes. Learn moreBy default, every PR changing files present in the Lunaria configuration's You can change this by adding one of the keywords present in the Tracked Files
Warnings reference
|
Co-authored-by: Daniel Roe <daniel@roe.dev>
Co-authored-by: Anthony Fu <github@antfu.me>
Co-authored-by: Alec Lloyd Probert <55991794+graphieros@users.noreply.github.com> Co-authored-by: autofix-ci[bot] <114827586+autofix-ci[bot]@users.noreply.github.com>
Co-authored-by: Willow (GHOST) <git@willow.sh>
Uh oh!
There was an error while loading. Please reload this page.
This PR will deploy the following changes to production (
npmx.dev).Next version:
v0.3.0(current:v0.2.2)Features
d10e3c36)e42fff4a)6a5e6b21)225b6d45)Fixes
58da5975)ee3ffaa0)4f4b600a)55308739)71116e8f)bf12be06)f9c8e9c8)summary(fix(lunaria): underline anchor tags and lighten consistently on hover likesummary#1937) (e69428d7)a682a99c)51ecbc77)11c2116d)01abe3b2)e3209f2f)af71229f)7c249c59)462fb984)6a44d718)9c5a6cb6)9216309a)f2045e66)d87e85c5)Other Changes
6a78b90e)f06b73f0)vue-data-uicss in components that use it (perf: only loadvue-data-uicss in components that use it #1918) (f3403eb5)resolveDependencyTree(test: add coverage for criticalresolveDependencyTree#1911) (94f95fad)036dfc2c)