Uh oh!
There was an error while loading. Please reload this page.
chore(deps): update all non-major dependencies - #2219
Conversation
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
Codecov Report✅ All modified and coverable lines are covered by tests. 📢 Thoughts on this report? Let us know! |
4a01cd8 to
91d0cc4Compare91d0cc4 to
e050b87Comparee050b87 to
a4b0b7fComparea4b0b7f to
7652e6cCompareghostdevv
commented
Mar 23, 2026
I'm not sure what's causing the issue, will ask if anyone knows in the discord o7 |
7652e6c to
da3b775Compareda3b775 to
f7645c9Comparef7645c9 to
a0ad39bComparea0ad39b to
a3903b9Compare3e64d4a to
5b370bbCompare5b370bb to
123e723Compare123e723 to
de5fa44Comparede5fa44 to
dcce6c5CompareEdited/Blocked NotificationRenovate will not automatically rebase this PR, because it does not recognize the last commit author and assumes somebody else may have edited the PR. You can manually request rebase by checking the rebase/retry box above. |
43081j
commented
Apr 6, 2026
FYI the bsky team broke the lex client which is why this doesn't build. once they fix it on their end, it will unblock this PR. |
ghostdevv
commented
Apr 7, 2026
yea I figured as much 🙃 - I pushed to stop this rebasing every two minutes xD |
📊 Dependency Size ChangesWarning This PR adds 14.9 MB of new dependencies, which exceeds the threshold of 200 kB.
Total size change: 14.9 MB |
Review the following changes in direct dependencies. Learn more about Socket for GitHub. |
edit: bot message was out of date so deleted it