Skip to content

chore: release v0.14.0 - #2810

Merged
danielroe merged 24 commits into
releasefrom
main
May 31, 2026
Merged

chore: release v0.14.0#2810
danielroe merged 24 commits into
releasefrom
main

Conversation

@github-actions

@github-actionsgithub-actionsBot commented May 28, 2026

Copy link
Copy Markdown

This PR will deploy the following changes to production (npmx.dev).

Next version: v0.14.0 (current: v0.13.0)

Features

Fixes

Other Changes


Merging this PR will:

  • Deploy to npmx.dev via Vercel
  • Create a v0.14.0 tag and GitHub Release
  • Publish npmx-connector@0.14.0 to npm

@github-actions
github-actionsBot requested a review from a teamMay 28, 2026 05:56
@github-actionsgithub-actionsBot added the release Production release PR label May 28, 2026
@github-actions
github-actionsBot requested a review from a teamMay 28, 2026 05:56
@vercel

vercelBot commented May 28, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
npmx-lunariaReadyReadyPreview, CommentMay 31, 2026 7:02pm
npmx.devReadyReadyPreview, CommentMay 31, 2026 7:02pm
1 Skipped Deployment
ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devIgnoredIgnoredPreviewMay 31, 2026 7:02pm

Request Review

@github-actions

github-actionsBot commented May 28, 2026

Copy link
Copy Markdown
Author

Lunaria Status Overview

🌕 This pull request will trigger status changes.

Learn more

By default, every PR changing files present in the Lunaria configuration's files property will be considered and trigger status changes accordingly.

You can change this by adding one of the keywords present in the ignoreKeywords property in your Lunaria configuration file in the PR's title (ignoring all files) or by including a tracker directive in the merged commit's description.

Tracked Files

FileNote
i18n/locales/ar.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/az-AZ.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/bg-BG.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/bn-IN.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/cs-CZ.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/de.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/en.jsonSource changed, localizations will be marked as outdated.
i18n/locales/es.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/fr-FR.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/hi-IN.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/hu-HU.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/id-ID.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/ja-JP.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/mr-IN.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/nb-NO.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/nl.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/pl-PL.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/pt-BR.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/pt-PT.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/ro-RO.jsonLocalization added, will be marked as complete. 🔄️
i18n/locales/ru-RU.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/sr-Latn-RS.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/tr-TR.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/uk-UA.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/vi-VN.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/zh-CN.jsonLocalization changed, will be marked as complete. 🔄️
i18n/locales/zh-TW.jsonLocalization changed, will be marked as complete. 🔄️
Warnings reference
IconDescription
🔄️The source for this localization has been updated since the creation of this pull request, make sure all changes in the source have been applied.

@codecov

codecovBot commented May 28, 2026

Copy link
Copy Markdown

@github-actions

github-actionsBot commented May 28, 2026

Copy link
Copy Markdown
Author

📊 Dependency Size Changes

Warning

This PR adds 12.2 MB of new dependencies, which exceeds the threshold of 200 kB.

📦 Package📏 Size
@rolldown/binding-linux-x64-gnu@1.0.0-rc.1223 MB
lightningcss-linux-x64-gnu@1.31.1-10 MB
@takumi-rs/core-linux-x64-gnu@0.73.17.4 MB
satori@0.19.2-5.3 MB
@shuding/opentype.js@1.4.0-beta.0-3 MB
valibot@1.3.0-1.8 MB
jiti@2.6.1-1.7 MB
reka-ui@2.8.2 → reka-ui@2.9.31.6 MB
@atproto/lex-schema@0.0.181.4 MB
@atproto/lex-schema@0.1.21.4 MB
@atproto/lex-schema@0.0.14-1.4 MB
@atproto/lexicon@0.6.21.1 MB
@atproto/lexicon@0.6.1-1.1 MB
@atproto/lexicon@0.7.11.1 MB
motion-v@1.10.3 → motion-v@2.2.1-799.7 kB
rolldown@1.0.0-rc.12759.5 kB
yaml@2.8.2-683 kB
@atproto/lex-client@0.1.3620.5 kB
@atproto/lex-client@0.0.19546.9 kB
fflate@0.7.4-535.4 kB
h3@1.15.8-529.1 kB
lightningcss@1.31.1-508.3 kB
nuxt-site-config@3.2.21-463.4 kB
cborg@1.10.2 → cborg@4.5.8-435.5 kB
@atproto/lex-client@0.0.15-435.4 kB
multiformats@13.4.2407.2 kB
@atproto/lex-data@0.1.1382.3 kB
@atproto/lex-data@0.0.15382 kB
@atproto/lex-data@0.0.13-353.4 kB
@atproto/lex-data@0.0.12-350.6 kB
htmlparser2@8.0.2-265 kB
@atproto/lex-json@0.0.16252.9 kB
@atproto/lex-json@0.1.0252.5 kB
yoga-layout@3.2.1-224.3 kB
linebreak@1.1.0-218.8 kB
@internationalized/date@3.11.0 → @internationalized/date@3.12.2192 kB
@oxc-resolver/binding-linux-x64-gnu@11.19.1 → @oxc-resolver/binding-linux-x64-gnu@11.20.0-188.4 kB
@nuxtjs/mdc@0.21.1182.4 kB
uint8arrays@5.1.1171.3 kB
prosemirror-markdown@1.13.4-163 kB
@atproto/common-web@0.4.21159.5 kB
get-tsconfig@4.13.7-156.6 kB
@atproto/common-web@0.4.17-155.4 kB
@atproto/common-web@0.5.0154.6 kB
@atproto/did@0.5.0151.4 kB
prosemirror-menu@1.3.0-141.4 kB
@atproto/xrpc@0.8.0140.2 kB
@atproto/api@0.19.0 → @atproto/api@0.20.7133.6 kB
@atproto/lex-cbor@0.0.14-127.9 kB
@atproto/lex-cbor@0.0.13-127.6 kB
framer-motion@12.34.3 → framer-motion@12.40.0126.6 kB
@atproto-labs/fetch@0.3.0125 kB
@tanstack/virtual-core@3.13.19 → @tanstack/virtual-core@3.16.0124.4 kB
motion-dom@12.34.3 → motion-dom@12.40.0124 kB
enhanced-resolve@5.20.0 → enhanced-resolve@5.22.1119.2 kB
@tiptap/markdown@3.20.0 → @tiptap/markdown@3.24.0118.4 kB
@atproto/lex-json@0.0.15110.6 kB
@nuxt/ui@4.5.1 → @nuxt/ui@4.6.1108.8 kB
@tiptap/extensions@3.20.0 → @tiptap/extensions@3.24.0107.1 kB
@remirror/core-constants@3.0.0-106.5 kB
@atproto/lex-json@0.0.13-104.2 kB
@tailwindcss/oxide-linux-x64-gnu@4.2.1 → @tailwindcss/oxide-linux-x64-gnu@4.3.0-104 kB
@atproto/lex-json@0.0.12-103.9 kB
@nuxt/kit@4.4.6101.6 kB
@tiptap/core@3.20.0 → @tiptap/core@3.24.0101.3 kB
semver@7.8.1100.2 kB
@nuxtjs/robots@5.7.1 → @nuxtjs/robots@6.0.9-91.9 kB
css-to-react-native@3.2.0-90 kB
@atproto/syntax@0.4.3 → @atproto/syntax@0.6.186.1 kB
@ai-sdk/gateway@3.0.66 → @ai-sdk/gateway@3.0.8382.7 kB
@nuxtjs/mcp-toolkit@0.7.0 → @nuxtjs/mcp-toolkit@0.13.480.2 kB
@shikijs/core@4.1.063.8 kB
@atproto-labs/did-resolver@0.3.160.9 kB
@tiptap/extension-drag-handle@3.20.0 → @tiptap/extension-drag-handle@3.24.055.5 kB
unicode-trie@2.0.0-48.9 kB
@ipld/dag-cbor@7.0.3-48.1 kB
@oxc-project/types@0.133.044.5 kB
@oxc-project/types@0.122.044.5 kB
@oxc-project/types@0.130.0-44.5 kB
css-gradient-parser@0.0.17-43.6 kB
@atproto/lex-cbor@0.1.043.3 kB
@tiptap/extension-list@3.20.0 → @tiptap/extension-list@3.24.041.3 kB
@shikijs/primitive@4.1.039.5 kB
@shikijs/transformers@4.1.037.9 kB
@takumi-rs/core@0.73.136.1 kB
@atproto-labs/simple-store@0.4.035.8 kB
@takumi-rs/helpers@0.73.134.6 kB
prosemirror-collab@1.3.1-32.5 kB
@shikijs/types@4.1.032.2 kB
prosemirror-schema-basic@1.2.4-30 kB
@atproto/lex@0.0.20 → @atproto/lex@0.1.329.9 kB
prosemirror-trailing-node@3.0.0-28.9 kB
nuxt-site-config-kit@3.2.21-26.3 kB
site-config-stack@3.2.21-25 kB
tapable@2.3.0 → tapable@2.3.325 kB
@atproto-labs/simple-store-memory@0.2.025 kB
parse-css-color@0.2.1-24 kB
@rolldown/pluginutils@1.0.0-rc.1223.6 kB
@oxc-parser/binding-linux-x64-gnu@0.130.0 → @oxc-parser/binding-linux-x64-gnu@0.133.020.5 kB
tailwindcss@4.2.1 → tailwindcss@4.3.0-20.2 kB
@babel/types@8.0.0-rc.2 → @babel/types@8.0.0-rc.3-19.2 kB
@modelcontextprotocol/sdk@1.27.1 → @modelcontextprotocol/sdk@1.29.018.5 kB
@tiptap/suggestion@3.20.0 → @tiptap/suggestion@3.24.017 kB
@tiptap/vue-3@3.20.0 → @tiptap/vue-3@3.24.016.6 kB
@atproto/crypto@0.4.5 → @atproto/crypto@0.5.0-15.2 kB
@atproto-labs/pipe@0.2.014 kB
virtua@0.48.8 → virtua@0.49.0-13 kB
@tiptap/extension-node-range@3.20.0 → @tiptap/extension-node-range@3.24.012.9 kB
camelize@1.0.1-12.5 kB
knip@6.14.2 → knip@6.15.011.9 kB
emoji-regex-xs@2.0.1-10.4 kB
@atproto/repo@0.8.12 → @atproto/repo@0.10.0-10 kB
docus@5.8.1 → docus@5.9.09.9 kB
packumeta@0.4.18.3 kB
@atproto/lex-builder@0.0.17 → @atproto/lex-builder@0.1.28.2 kB
ai@6.0.116 → ai@6.0.1417.6 kB
algoliasearch@5.49.2 → algoliasearch@5.50.17.3 kB
axe-core@4.11.1 → axe-core@4.11.27 kB
hex-rgb@4.3.0-6.7 kB
@babel/parser@8.0.0-rc.2 → @babel/parser@8.0.0-rc.3-6.6 kB
tsdown@0.21.4 → tsdown@0.21.76.5 kB
@atproto/lex-resolver@0.0.17 → @atproto/lex-resolver@0.1.06.1 kB
@tiptap/extension-drag-handle-vue-3@3.20.0 → @tiptap/extension-drag-handle-vue-3@3.24.06 kB
vue-component-type-helpers@3.3.35.6 kB
vue-component-type-helpers@3.3.1-5.6 kB
vue-component-type-helpers@3.2.8-5.6 kB
crelt@1.0.6-5.4 kB
string.prototype.codepointat@0.2.1-5.4 kB
@algolia/client-common@5.49.2 → @algolia/client-common@5.50.15.1 kB
@atproto/lex-installer@0.0.20 → @atproto/lex-installer@0.1.05 kB
@tiptap/extension-paragraph@3.20.0 → @tiptap/extension-paragraph@3.24.04.2 kB
@iconify-json/lucide@1.2.98 → @iconify-json/lucide@1.2.1014 kB
oxc-parser@0.130.0 → oxc-parser@0.133.03.8 kB
@tiptap/extension-bubble-menu@3.20.0 → @tiptap/extension-bubble-menu@3.24.03.8 kB
@algolia/abtesting@1.15.2 → @algolia/abtesting@1.16.13.6 kB
@algolia/client-abtesting@5.49.2 → @algolia/client-abtesting@5.50.13.6 kB
@algolia/client-analytics@5.49.2 → @algolia/client-analytics@5.50.13.6 kB
@algolia/client-insights@5.49.2 → @algolia/client-insights@5.50.13.6 kB
@algolia/client-personalization@5.49.2 → @algolia/client-personalization@5.50.13.6 kB
@algolia/client-query-suggestions@5.49.2 → @algolia/client-query-suggestions@5.50.13.6 kB
@algolia/ingestion@1.49.2 → @algolia/ingestion@1.50.13.6 kB
@algolia/monitoring@1.49.2 → @algolia/monitoring@1.50.13.6 kB
@algolia/client-search@5.49.2 → @algolia/client-search@5.50.13.6 kB
@algolia/recommend@5.49.2 → @algolia/recommend@5.50.13.6 kB
@iconify-json/simple-icons@1.2.74 → @iconify-json/simple-icons@1.2.763.4 kB
@tiptap/extension-floating-menu@3.20.0 → @tiptap/extension-floating-menu@3.24.03.3 kB
@tiptap/pm@3.20.0 → @tiptap/pm@3.24.0-3.1 kB
@atproto/lex-password-session@0.0.8 → @atproto/lex-password-session@0.0.123 kB
rolldown-plugin-dts@0.22.5 → rolldown-plugin-dts@0.23.22.4 kB
@tiptap/extension-collaboration@3.20.0 → @tiptap/extension-collaboration@3.24.0-2.1 kB
@tailwindcss/node@4.2.1 → @tailwindcss/node@4.3.02.1 kB
@tiptap/extension-link@3.20.0 → @tiptap/extension-link@3.24.02 kB
@ai-sdk/vue@3.0.116 → @ai-sdk/vue@3.0.1411.7 kB
@tailwindcss/postcss@4.2.1 → @tailwindcss/postcss@4.3.01.7 kB
@atproto/lex-document@0.0.15 → @atproto/lex-document@0.1.0-1.6 kB
motion-utils@12.29.2 → motion-utils@12.39.0-1.5 kB
await-lock@2.2.2 → await-lock@3.0.0-1.4 kB
@ai-sdk/provider-utils@4.0.19 → @ai-sdk/provider-utils@4.0.211.2 kB
@algolia/requester-node-http@5.49.2 → @algolia/requester-node-http@5.50.11.2 kB
unplugin-vue-components@31.0.0 → unplugin-vue-components@32.1.0-1.1 kB
@atproto/common@0.5.13 → @atproto/common@0.6.11.1 kB
@lydell/node-pty@1.2.0-beta.3 → @lydell/node-pty@1.2.0-beta.121.1 kB
@tiptap/extension-blockquote@3.20.0 → @tiptap/extension-blockquote@3.24.0686 B
@babel/helper-string-parser@8.0.0-rc.2 → @babel/helper-string-parser@8.0.0-rc.6-682 B
@babel/helper-validator-identifier@8.0.0-rc.2 → @babel/helper-validator-identifier@8.0.0-rc.3660 B
@tailwindcss/vite@4.2.1 → @tailwindcss/vite@4.3.0591 B
defu@6.1.4 → defu@6.1.6587 B
@tiptap/extension-code-block@3.20.0 → @tiptap/extension-code-block@3.24.0583 B
zod-to-json-schema@3.25.1 → zod-to-json-schema@3.25.2582 B
@tiptap/extension-bold@3.20.0 → @tiptap/extension-bold@3.24.0577 B
@tiptap/extension-italic@3.20.0 → @tiptap/extension-italic@3.24.0537 B
sanitize-html@2.17.1 → sanitize-html@2.17.2424 B
@babel/generator@8.0.0-rc.2 → @babel/generator@8.0.0-rc.3418 B
@lydell/node-pty-linux-x64@1.2.0-beta.3 → @lydell/node-pty-linux-x64@1.2.0-beta.12363 B
linkifyjs@4.3.2 → linkifyjs@4.3.3-351 B
vue-data-ui@3.20.10 → vue-data-ui@3.20.11338 B
chromatic@16.0.0 → chromatic@16.1.0-306 B
@tiptap/extension-mention@3.20.0 → @tiptap/extension-mention@3.24.0198 B
@tiptap/extension-hard-break@3.20.0 → @tiptap/extension-hard-break@3.24.0141 B
@tailwindcss/oxide@4.2.1 → @tailwindcss/oxide@4.3.0130 B
@tiptap/extension-heading@3.20.0 → @tiptap/extension-heading@3.24.0-114 B
@tiptap/extension-code@3.20.0 → @tiptap/extension-code@3.24.0-105 B
@tiptap/extension-image@3.20.0 → @tiptap/extension-image@3.24.0-105 B
@tiptap/extension-placeholder@3.20.0 → @tiptap/extension-placeholder@3.24.0-105 B
@tiptap/extension-bullet-list@3.20.0 → @tiptap/extension-bullet-list@3.24.0-105 B
@tiptap/extension-document@3.20.0 → @tiptap/extension-document@3.24.0-105 B
@tiptap/extension-dropcursor@3.20.0 → @tiptap/extension-dropcursor@3.24.0-105 B
@tiptap/extension-gapcursor@3.20.0 → @tiptap/extension-gapcursor@3.24.0-105 B
@tiptap/extension-list-item@3.20.0 → @tiptap/extension-list-item@3.24.0-105 B
@tiptap/extension-list-keymap@3.20.0 → @tiptap/extension-list-keymap@3.24.0-105 B
@tiptap/extension-ordered-list@3.20.0 → @tiptap/extension-ordered-list@3.24.0-105 B
@tiptap/extension-strike@3.20.0 → @tiptap/extension-strike@3.24.0-105 B
@tiptap/extension-text@3.20.0 → @tiptap/extension-text@3.24.0-105 B
@tiptap/extension-underline@3.20.0 → @tiptap/extension-underline@3.24.0-105 B
@tiptap/starter-kit@3.20.0 → @tiptap/starter-kit@3.24.0-67 B
@algolia/requester-browser-xhr@5.49.2 → @algolia/requester-browser-xhr@5.50.145 B
@algolia/requester-fetch@5.49.2 → @algolia/requester-fetch@5.50.144 B
@tiptap/extension-horizontal-rule@3.20.0 → @tiptap/extension-horizontal-rule@3.24.0-36 B
oxc-resolver@11.19.1 → oxc-resolver@11.20.0-16 B
@tanstack/vue-virtual@3.13.19 → @tanstack/vue-virtual@3.13.26-1 B
local-pkg@1.1.2 → local-pkg@1.2.10 B
base64-js@0.0.8Unknown
css-color-keywords@1.0.0Unknown
pako@0.2.9Unknown
css-background-parser@0.1.0Unknown
css-box-shadow@1.0.0-3Unknown

Total size change: 12.2 MB

Co-authored-by: autofix-ci[bot] <114827586+autofix-ci[bot]@users.noreply.github.com>
Co-authored-by: Willow (GHOST) <git@willow.sh>
…2792)
Co-authored-by: Roman <dev@rman.dev>
Co-authored-by: Willow (GHOST) <git@willow.sh>
Co-authored-by: Nicolas Charpentier <7189823+charpeni@users.noreply.github.com>
Co-authored-by: ProCode <68168458+ProCode-Software@users.noreply.github.com>
Co-authored-by: WenLong Chen <50607756+stellhub@users.noreply.github.com>
Co-authored-by: Willow (GHOST) <git@willow.sh>
Co-authored-by: autofix-ci[bot] <114827586+autofix-ci[bot]@users.noreply.github.com>
cylewaitforitand others added 2 commits May 31, 2026 15:55
Co-authored-by: Willow (GHOST) <git@willow.sh>
Co-authored-by: autofix-ci[bot] <114827586+autofix-ci[bot]@users.noreply.github.com>
Co-authored-by: Roman <dev@rman.dev>
Co-authored-by: Felix Schneider <99918022+trueberryless@users.noreply.github.com>
Co-authored-by: Willow (GHOST) <git@willow.sh>
Co-authored-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
Co-authored-by: Willow (GHOST) <git@willow.sh>
Co-authored-by: James Garbutt <43081j@users.noreply.github.com>
Co-authored-by: Philippe Serhal <philippe.serhal@gmail.com>
@danielroe
danielroe merged commit 36128a5 into releaseMay 31, 2026
55 of 58 checks passed
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

releaseProduction release PR

Projects

None yet

Development

Successfully merging this pull request may close these issues.

20 participants

@patak-cat@ghostdevv@danielroe@mihaizaurus@rohitg00@johnnyreilly@43081j@graphieros@alexdln@tinas@shomriddho@btea@serhalp@jp-knj@charpeni@gameroman@cylewaitforit@OrbisK@shuuji3@trueberryless