Skip to content

chore: fix workflow-open action - #2265

Closed
ghostdevv wants to merge 1 commit into
mainfrom
g/welcome
Closed

chore: fix workflow-open action#2265
ghostdevv wants to merge 1 commit into
mainfrom
g/welcome

Conversation

@ghostdevv

Copy link
Copy Markdown
Member

🔗 Linked issue

n/a

🧭 Context

The action we added in #2236 is broked

📚 Description

@vercel

vercelBot commented Mar 24, 2026

Copy link
Copy Markdown
Contributor

The latest updates on your projects. Learn more about Vercel for GitHub.

ProjectDeploymentActionsUpdated (UTC)
npmx.devReadyReadyPreview, CommentMar 24, 2026 3:22pm
2 Skipped Deployments
ProjectDeploymentActionsUpdated (UTC)
docs.npmx.devIgnoredIgnoredMar 24, 2026 3:22pm
npmx-lunariaIgnoredIgnoredMar 24, 2026 3:22pm

Request Review

@ghostdevv

Copy link
Copy Markdown
MemberAuthor

cc: @trueberryless

@ghostdevv

Copy link
Copy Markdown
MemberAuthor

ah! #2264

@codecov

codecovBot commented Mar 24, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ All tests successful. No failed tests found.

📢 Thoughts on this report? Let us know!

@ghostdevv
ghostdevv deleted the g/welcome branch April 4, 2026 19:24
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@ghostdevv