Skip to content

AppFrame component - #2147

Merged
vdepizzol merged 5 commits into
mainfrom
vd/AppFrame
Jul 1, 2022
Merged

AppFrame component#2147
vdepizzol merged 5 commits into
mainfrom
vd/AppFrame

Conversation

@vdepizzol

@vdepizzolvdepizzol commented Jul 1, 2022

Copy link
Copy Markdown
Contributor

AppFrame is a thin wrapper that holds the page together, while making sure that the page's global footer is always visible "after the fold". This guarantees that loading states won't show the footer temporarily, and allows the footer to include richer interactions without obfuscating or competing with the page contents.

Features

  • Accessible navigation with "Skip to content"-style anchors
  • Unopinionated generic slots/regions for:
    • header
    • subheader
    • body
    • footer

Can these changes ship as is?

  • Yes, this PR does not depend on additional changes. 🚢

@vdepizzol
vdepizzol requested a review from a team as a code ownerJuly 1, 2022 18:35
@vdepizzol
vdepizzol requested a review from langermankJuly 1, 2022 18:35
@changeset-bot

changeset-botBot commented Jul 1, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 67efaa3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vdepizzol
vdepizzol temporarily deployed to github-pages July 1, 2022 18:40 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 1, 2022 18:41 Inactive
@vdepizzol
vdepizzol temporarily deployed to github-pages July 1, 2022 19:00 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 1, 2022 19:01 Inactive

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

One tiny comment on the missing fallback, otherwise looks good!

Comment threadsrc/layout/app-frame.scss Outdated
Comment threadsrc/layout/app-frame.scss Outdated
Comment threadsrc/layout/app-frame.scss
vdepizzoland others added 2 commits July 1, 2022 12:33
Co-authored-by: Katie Langerman <langermank@github.com>
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 1, 2022 19:40 Inactive
@vdepizzol
vdepizzol temporarily deployed to github-pages July 1, 2022 19:44 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 1, 2022 19:44 Inactive
@github-actions
github-actionsBot temporarily deployed to Storybook Preview July 1, 2022 19:54 Inactive
@vdepizzol
vdepizzol merged commit 9dd2a49 into mainJul 1, 2022
@vdepizzol
vdepizzol deleted the vd/AppFrame branch July 1, 2022 20:00
@primer-cssprimer-css mentioned this pull request Jul 1, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@vdepizzol@langermank