Skip to content

[Major] Remove legacy color support - #2607

Closed
langermank wants to merge 9 commits into
mainfrom
remove-old-colors
Closed

[Major] Remove legacy color support#2607
langermank wants to merge 9 commits into
mainfrom
remove-old-colors

Conversation

@langermank

@langermanklangermank commented Apr 9, 2024

Copy link
Copy Markdown
Contributor

This PR removes support for color themes. Going forward, Primer expects an explicit color theme import directly from primer/primitives.

@import '@primer/primitives/dist/css/functional/themes/light.css';
@import '@primer/primitives/dist/css/functional/themes/light-tritanopia.css';
@import '@primer/primitives/dist/css/functional/themes/light-high-contrast.css';
@import '@primer/primitives/dist/css/functional/themes/light-colorblind.css';
@import '@primer/primitives/dist/css/functional/themes/dark.css';
@import '@primer/primitives/dist/css/functional/themes/dark-colorblind.css';
@import '@primer/primitives/dist/css/functional/themes/dark-dimmed.css';
@import '@primer/primitives/dist/css/functional/themes/dark-high-contrast.css';
@import '@primer/primitives/dist/css/functional/themes/dark-tritanopia.css';

This is a breaking change and marked as such.

@changeset-bot

changeset-botBot commented Apr 9, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 77ef79f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/cssMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment threadsrc/color-modes/themes/dark.scss Outdated
@include color-mode-theme(dark) {
@include primer-colors-dark;
}
@import '@primer/primitives/dist/css/functional/themes/dark.css';

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

If we're gonna go with a major, I say we just completely remove the src/color-modes/ folder and use the primitives imports directly in dotcom

Copy link
Copy Markdown
ContributorAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🔥

@github-actions
github-actionsBottemporarily deployed to Storybook Preview May 10, 2024 17:45 Inactive
@langermanklangermank changed the title [Draft] next major[Major] Remove legacy color supportMay 16, 2024
@langermank
langermank marked this pull request as ready for review May 16, 2024 21:32
@langermank
langermank requested review from a team as code ownersMay 16, 2024 21:32
@langermank
langermank requested a review from mperrottiMay 16, 2024 21:32
@github-actions

Copy link
Copy Markdown
Contributor

Hi! This pull request has been marked as stale because it has been open with no activity for 60 days. You can comment on the pull request or remove the stale label to keep it open. If you do nothing, this pull request will be closed in 7 days.

@github-actionsgithub-actionsBot added the Stale Automatically marked as stale. label Oct 11, 2024
@github-actions
github-actionsBot deleted the remove-old-colors branch October 18, 2024 08:07
@langermank
langermank restored the remove-old-colors branch February 4, 2025 21:05
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

major releaseStaleAutomatically marked as stale.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@jonrohan