Skip to content

ActionList: Better label & description for Item - #1585

Closed
siddharthkp wants to merge 114 commits into
mainfrom
siddharth/action-list-description-a11y
Closed

ActionList: Better label & description for Item#1585
siddharthkp wants to merge 114 commits into
mainfrom
siddharth/action-list-description-a11y

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 8, 2021

Copy link
Copy Markdown
Member

Change base after #1517 is merged
Part of https://github.com/github/primer/issues/447

  • Bring inline description to be part of label.
  • Block description is describedby

This is the generated HTML, formatted for legibility:

image

<ul><liaria-labelledby="react-aria-1 react-aria-2"><span><imgsrc="https://github.com/pksjce.png"/></span><divdata-component="ActionList.Item--DividerContainer"><div><spanid="react-aria-1">pksjce</span><divid="react-aria-2" title="Pavithra Kodmad">Pavithra Kodmad</div></div></div></li><li>...</li><lirole="separator"></li><liaria-labelledby="react-aria-3 " aria-describedby="react-aria-4"><span><spanclass="label-circle"></span></span><divdata-component="ActionList.Item--DividerContainer"><spanid="react-aria-3">blocked</span><spanid="react-aria-4">
Someone or something is preventing this from moving forward
</span></div></li><li>...</li></ul>

Merge checklist

@changeset-bot

changeset-botBot commented Nov 8, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 08e944c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp changed the title bring inline description to labelActionList: Better label & description for ItemNov 8, 2021
@github-actions

github-actionsBot commented Nov 8, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js54.42 KB (0%)
dist/browser.umd.js54.81 KB (0%)

@siddharthkpsiddharthkp self-assigned this Nov 8, 2021
Base automatically changed from action-list-2 to mainNovember 11, 2021 13:32
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

Moved to #1599

@joshblack
joshblack deleted the siddharth/action-list-description-a11y branch January 19, 2023 16:44
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@jfuchs