Skip to content

ActionList: Better label & description for Item - #1599

Merged
siddharthkp merged 2 commits into
mainfrom
siddharth/actionlist-description-a11y
Nov 15, 2021
Merged

ActionList: Better label & description for Item#1599
siddharthkp merged 2 commits into
mainfrom
siddharth/actionlist-description-a11y

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Nov 11, 2021

Copy link
Copy Markdown
Member

Change base after #1517 is merged
Part of https://github.com/github/primer/issues/447

  • Bring inline description to be part of label.
  • Block description is describedby

This is the generated HTML, formatted for legibility:

image

<ul><liaria-labelledby="react-aria-1 react-aria-2"><span><imgsrc="https://github.com/pksjce.png"/></span><divdata-component="ActionList.Item--DividerContainer"><div><spanid="react-aria-1">pksjce</span><divid="react-aria-2" title="Pavithra Kodmad">Pavithra Kodmad</div></div></div></li><li>...</li><lirole="separator"></li><liaria-labelledby="react-aria-3 " aria-describedby="react-aria-4"><span><spanclass="label-circle"></span></span><divdata-component="ActionList.Item--DividerContainer"><spanid="react-aria-3">blocked</span><spanid="react-aria-4">
Someone or something is preventing this from moving forward
</span></div></li><li>...</li></ul>

Merge checklist

@siddharthkp
siddharthkp requested review from a team and jfuchsNovember 11, 2021 15:43
@changeset-bot

changeset-botBot commented Nov 11, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2be0740

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/componentsPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 11, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js54.44 KB (0%)
dist/browser.umd.js54.83 KB (0%)

@siddharthkp
siddharthkp merged commit a2e195b into mainNov 15, 2021
@siddharthkp
siddharthkp deleted the siddharth/actionlist-description-a11y branch November 15, 2021 13:55
This was referenced Nov 15, 2021
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@pksjce