Skip to content

Button v2 - Fix font family - #1734

Merged
pksjce merged 2 commits into
mainfrom
siddharth/newbutton-font
Dec 15, 2021
Merged

Button v2 - Fix font family#1734
pksjce merged 2 commits into
mainfrom
siddharth/newbutton-font

Conversation

@siddharthkp

Copy link
Copy Markdown
Member

Describe your changes here.

Button v1 has font-family: inherit, this was missing from NewButton which made it use Arial on Chrome instead of system-font

Screenshots

before:
Before fix, Button has Arial

after:
Before fix, Button has system font

@siddharthkp
siddharthkp requested review from a team and pksjceDecember 14, 2021 18:22
@changeset-bot

changeset-botBot commented Dec 14, 2021

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 35a8c44

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp changed the title Add font inherit to NewButtonNewButton - Fix font familyDec 14, 2021
@siddharthkpsiddharthkp self-assigned this Dec 14, 2021
@siddharthkpsiddharthkp changed the title NewButton - Fix font familyButton v2 - Fix font familyDec 14, 2021
@siddharthkpsiddharthkp changed the title Button v2 - Fix font familyButton2 - Fix font familyDec 14, 2021
@siddharthkpsiddharthkp changed the title Button2 - Fix font familyButton v2 - Fix font familyDec 14, 2021
@github-actions

github-actionsBot commented Dec 14, 2021

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js57.27 KB (+0.01% 🔺)
dist/browser.umd.js57.59 KB (+0.01% 🔺)

@pksjcepksjce left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks!

@pksjce
pksjce merged commit bba4c24 into mainDec 15, 2021
@pksjce
pksjce deleted the siddharth/newbutton-font branch December 15, 2021 05:00
@primer-cssprimer-css mentioned this pull request Dec 15, 2021
pksjce pushed a commit that referenced this pull request Dec 20, 2021
* Add font inherit to NewButton
* Create tiny-cows-fold.md
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@siddharthkp@pksjce@colebemis