Skip to content

Fix ActionList group roles - #1851

Closed
siddharthkp wants to merge 18 commits into
mainfrom
siddharth/actionlist-group-role
Closed

Fix ActionList group roles#1851
siddharthkp wants to merge 18 commits into
mainfrom
siddharth/actionlist-group-role

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 9, 2022

Copy link
Copy Markdown
Member

Before:

<ulrole="none"><li><divrole="presentation" aria-hidden="true" id="1">Open</div><ulrole="menu" aria-labelledby="1"><lirole="menuitemradio">FY22 - Q3</li><lirole="menuitemradio">FY22 - Q2</li></ul></li><li><divrole="presentation" aria-hidden="true" id="2">Closed</div><ulrole="menu" aria-labelledby="2"><lirole="menuitemradio">FY22 - Q1</li><lirole="menuitemradio">FY21 - Q4</li></ul></li></ul>

Accessibility tree:

generic
generic
menu "Open"
menuitemradio "FY22 - Q3"
menuitemradio "FY22 - Q2"
menu "Closed"
menuitemradio "FY22 - Q1"
menuitemradio "FY21 - Q4"

After:

<ulrole="menu"><lirole="none> <div role="presentation" aria-hidden="true" id="1">Open</div><ulrole="group" aria-labelledby="1"><lirole="menuitemradio">FY22 - Q3</li><lirole="menuitemradio">FY22 - Q2</li></ul></li><lirole="menu"><divrole="presentation" aria-hidden="true" id="2">Closed</div><ulrole="group" aria-labelledby="2"><lirole="menuitemradio">FY22 - Q1</li><lirole="menuitemradio">FY21 - Q4</li></ul></li></ul>

Accessibility tree:

menu
group "Open"
menuitemradio "FY22 - Q3"
menuitemradio "FY22 - Q2"
group "Closed"
menuitemradio "FY22 - Q1"
menuitemradio "FY21 - Q4"

@changeset-bot

changeset-botBot commented Feb 9, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3a8162d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 9, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.5 KB (0%)
dist/browser.umd.js62.88 KB (0%)

@rezrah
rezrah self-requested a review February 14, 2022 16:41
siddharthkp added a commit that referenced this pull request Feb 21, 2022
@siddharthkp

Copy link
Copy Markdown
MemberAuthor

Replaced by #1876

siddharthkp added a commit that referenced this pull request Feb 23, 2022
* Bring in changes from #1851
* wasn't supposed to commit that
* Fix ActionMenu stories
* Fix ActionList/Groups story
* Narrow role for listRole to AriaRole
@joshblack
joshblack deleted the siddharth/actionlist-group-role branch January 19, 2023 16:47
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@rezrah