Skip to content

Fix ActionList group roles - #1876

Merged
siddharthkp merged 6 commits into
mainfrom
siddharth/actionlist-group-a11y
Feb 23, 2022
Merged

Fix ActionList group roles#1876
siddharthkp merged 6 commits into
mainfrom
siddharth/actionlist-group-a11y

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Feb 21, 2022

Copy link
Copy Markdown
Member

Before:

<ulrole="none"><li><divrole="presentation" aria-hidden="true" id="1">Open</div><ulrole="menu" aria-labelledby="1"><lirole="menuitemradio">FY22 - Q3</li><lirole="menuitemradio">FY22 - Q2</li></ul></li><li><divrole="presentation" aria-hidden="true" id="2">Closed</div><ulrole="menu" aria-labelledby="2"><lirole="menuitemradio">FY22 - Q1</li><lirole="menuitemradio">FY21 - Q4</li></ul></li></ul>

Accessibility tree:

generic
generic
menu "Open"
menuitemradio "FY22 - Q3"
menuitemradio "FY22 - Q2"
menu "Closed"
menuitemradio "FY22 - Q1"
menuitemradio "FY21 - Q4"

After:

<ulrole="menu"><lirole="none> <div role="presentation" aria-hidden="true" id="1">Open</div><ulrole="group" aria-labelledby="1"><lirole="menuitemradio">FY22 - Q3</li><lirole="menuitemradio">FY22 - Q2</li></ul></li><lirole="none"><divrole="presentation" aria-hidden="true" id="2">Closed</div><ulrole="group" aria-labelledby="2"><lirole="menuitemradio">FY22 - Q1</li><lirole="menuitemradio">FY21 - Q4</li></ul></li></ul>

Accessibility tree:

menu
group "Open"
menuitemradio "FY22 - Q3"
menuitemradio "FY22 - Q2"
group "Closed"
menuitemradio "FY22 - Q1"
menuitemradio "FY21 - Q4"

@changeset-bot

changeset-botBot commented Feb 21, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f98e53e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 21, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js62.62 KB (0%)
dist/browser.umd.js63.01 KB (0%)

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

:shipit:

@siddharthkp
siddharthkp enabled auto-merge (squash) February 23, 2022 11:38
@siddharthkp
siddharthkp merged commit 6cc9260 into mainFeb 23, 2022
@siddharthkp
siddharthkp deleted the siddharth/actionlist-group-a11y branch February 23, 2022 11:38
@primer-cssprimer-css mentioned this pull request Feb 23, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@siddharthkp@colebemis