Skip to content

Enable TextInputWithTokens to accept leading and trailing visuals - #1886

Merged
mperrotti merged 6 commits into
mainfrom
mp/textinputwithtokens-visuals
Feb 25, 2022
Merged

Enable TextInputWithTokens to accept leading and trailing visuals#1886
mperrotti merged 6 commits into
mainfrom
mp/textinputwithtokens-visuals

Conversation

@mperrotti

Copy link
Copy Markdown
Contributor

Makes it possible to render leading and trailing visuals in TextInputWithTokens just like we do in TextInput

Closes#1875

Screenshots

Screen Shot 2022-02-23 at 4 12 58 PM
Screen Shot 2022-02-23 at 4 13 12 PM

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@mperrotti
mperrotti requested review from a team and colebemisFebruary 23, 2022 21:15
@changeset-bot

changeset-botBot commented Feb 23, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b47f3d6

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 23, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js64.24 KB (+0.1% 🔺)
dist/browser.umd.js64.6 KB (+0.1% 🔺)

@mperrotti
mperrottiforce-pushed the mp/textinputwithtokens-visuals branch from 683bac0 to 892fe86CompareFebruary 23, 2022 21:19

@rezrahrezrah left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM.

Would be nice to get some examples in the docs?

@mperrotti
mperrotti merged commit ecbf923 into mainFeb 25, 2022
@mperrotti
mperrotti deleted the mp/textinputwithtokens-visuals branch February 25, 2022 23:17
This was referenced Feb 25, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

TextInputWithTokens does not render leading/trailingVisual

2 participants

@mperrotti@rezrah