Skip to content

[Storybook] Remove deprecated components in Overlay - #2555

Merged
josepmartins merged 6 commits into
mainfrom
josepmartins/overlay-stories-fixes
Nov 9, 2022
Merged

[Storybook] Remove deprecated components in Overlay#2555
josepmartins merged 6 commits into
mainfrom
josepmartins/overlay-stories-fixes

Conversation

@josepmartins

@josepmartinsjosepmartins commented Nov 8, 2022

Copy link
Copy Markdown
Contributor

Removing deprecated component usage as we plan to heavily invest in Storybook and Lookbook as the primary source for code documentation:

  • Remove deprecated Button variants and Dropdown instances in Overlay stories.
  • Removed styled-components

Related: https://github.com/github/primer/issues/1375

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@josepmartins
josepmartins requested review from a team and mperrottiNovember 8, 2022 15:04
@changeset-bot

changeset-botBot commented Nov 8, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7b03068

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@josepmartinsjosepmartins self-assigned this Nov 8, 2022
@github-actions

github-actionsBot commented Nov 8, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.87 KB (0%)
dist/browser.umd.js79.52 KB (0%)

@josepmartins
josepmartins temporarily deployed to github-pages November 8, 2022 15:12 Inactive
@josepmartins
josepmartins temporarily deployed to github-pages November 8, 2022 15:37 Inactive
@josepmartins
josepmartins temporarily deployed to github-pages November 8, 2022 16:34 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

💖

@josepmartins
josepmartins temporarily deployed to github-pages November 9, 2022 08:27 Inactive
@josepmartins
josepmartins merged commit dc1cf6d into mainNov 9, 2022
@josepmartins
josepmartins deleted the josepmartins/overlay-stories-fixes branch November 9, 2022 11:25
@primer-cssprimer-css mentioned this pull request Nov 9, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@josepmartins@colebemis