Skip to content

[Storybook] Remove deprecated components in stories - #2563

Merged
josepmartins merged 10 commits into
mainfrom
josepmartins/storybook-deprecated-components
Nov 11, 2022
Merged

[Storybook] Remove deprecated components in stories#2563
josepmartins merged 10 commits into
mainfrom
josepmartins/storybook-deprecated-components

Conversation

@josepmartins

@josepmartinsjosepmartins commented Nov 10, 2022

Copy link
Copy Markdown
Contributor

Remove deprecated component instances as we plan to heavily invest in Storybook and Lookbook as the primary source for code documentation.

Related: https://github.com/github/primer/issues/1375
Similar PR: #2555

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@josepmartins
josepmartins requested review from a team and broccolinisoupNovember 10, 2022 16:05
@changeset-bot

changeset-botBot commented Nov 10, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 20cea00

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 10, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.9 KB (0%)
dist/browser.umd.js79.54 KB (0%)

@josepmartinsjosepmartins self-assigned this Nov 10, 2022
@josepmartins
josepmartins temporarily deployed to github-pages November 10, 2022 16:21 Inactive

@langermanklangermank left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Wow, this is excellent! Great catch on this.

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🚀

@josepmartins
josepmartins temporarily deployed to github-pages November 11, 2022 08:02 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-2563 November 11, 2022 08:02 Inactive
@josepmartins
josepmartins merged commit 3085fde into mainNov 11, 2022
@josepmartins
josepmartins deleted the josepmartins/storybook-deprecated-components branch November 11, 2022 08:12
@primer-cssprimer-css mentioned this pull request Nov 11, 2022
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@josepmartins@broccolinisoup@langermank