Skip to content

UnderlineNav2: expose aria-current in the API and enable to use it for select status - #2618

Merged
broccolinisoup merged 9 commits into
mainfrom
UnderlineNav-aria-current
Nov 29, 2022
Merged

UnderlineNav2: expose aria-current in the API and enable to use it for select status#2618
broccolinisoup merged 9 commits into
mainfrom
UnderlineNav-aria-current

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Nov 27, 2022

Copy link
Copy Markdown
Member

This PR is mainly created for API consistency purposes between NavList and UnderlineNav v2.

UnderlineNav was using selected prop for the selected UnderlineNav.Item whereas NavList does the same thing with aria-current. This PR exposes aria-current attribute in the API for managing its selected state.

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@changeset-bot

changeset-botBot commented Nov 27, 2022

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0983b76

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Nov 27, 2022

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js78.92 KB (0%)
dist/browser.umd.js79.56 KB (0%)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 27, 2022 23:52 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 27, 2022 23:53 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 27, 2022 23:53 Inactive
@broccolinisoupbroccolinisoup changed the title UnderlineNav2: expose aria-current as an API and enable to use it for select statusUnderlineNav2: expose aria-current in the API and enable to use it for select statusNov 28, 2022
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 28, 2022 01:17 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 01:17 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 28, 2022 01:35 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 01:35 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 02:33 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 28, 2022 02:33 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 02:34 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review November 28, 2022 02:39
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 28, 2022 02:47 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 02:47 Inactive
@broccolinisoup
broccolinisoupforce-pushed the UnderlineNav-aria-current branch from 02407f6 to 0cafcd7CompareNovember 28, 2022 23:27
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 28, 2022 23:35 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 28, 2022 23:35 Inactive
@broccolinisoup
broccolinisoupforce-pushed the UnderlineNav-aria-current branch from 0cafcd7 to 989520bCompareNovember 29, 2022 00:15
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 29, 2022 00:23 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 29, 2022 00:23 Inactive

@colebemiscolebemis left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Nice work on this, @broccolinisoup 💖

Comment threadsrc/UnderlineNav2/UnderlineNav.tsx Outdated
Comment threadsrc/UnderlineNav2/UnderlineNavItem.tsx Outdated
Co-authored-by: Cole Bemis <colebemis@github.com>
@broccolinisoup
broccolinisoup temporarily deployed to github-pages November 29, 2022 03:58 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-2618 November 29, 2022 03:59 Inactive
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@broccolinisoup@colebemis