Uh oh!
There was an error while loading. Please reload this page.
PageLayout.Content should not use main landmark by default. - #3154
Conversation
🦋 Changeset detectedLatest commit: 4eb259c The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
size-limit report 📦
|
Uh oh!
There was an error while loading. Please reload this page.
TylerJDev
left a comment
There was a problem hiding this comment.
Very light feedback. Let me know your thoughts!
Pending your comment on role usage, we'll need to adjust the documentation in PageLayout.mdx for PageLayout.Content.
Uh oh!
There was an error while loading. Please reload this page.
Uh oh!
There was an error while loading. Please reload this page.
Co-authored-by: Tyler Jones <tylerjdev@github.com>
TylerJDev
left a comment
There was a problem hiding this comment.
Left a comment for role usage, but I think this is good to go 🔥! There's some wording we might have to change in the docs around main usage now - but this can be a follow up PR later as it isn't a blocker.
PageLayout.Contentwill no longer render asmainby default. Instead, end users can pass a specificasprop to set the rendered output.Closeshttps://github.com/github/primer/issues/1586
Screenshots
Please provide before/after screenshots for any visual changes
Merge checklist
Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.