Skip to content

Remove main landmark from PageLayout.Content. - #3675

Merged
radglob merged 2 commits into
mainfrom
remove-pagelayout-content-main-landmark
Aug 29, 2023
Merged

Remove main landmark from PageLayout.Content.#3675
radglob merged 2 commits into
mainfrom
remove-pagelayout-content-main-landmark

Conversation

@radglob

Copy link
Copy Markdown
Contributor

Reintroducing these changes which were introduced in #3154, but were reverted.

Closeshttps://github.com/github/primer/issues/1586

Screenshots

Please provide before/after screenshots for any visual changes

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@radglobradglob self-assigned this Aug 28, 2023
@radglob
radglob requested review from a team and joshblackAugust 28, 2023 15:24
@changeset-bot

changeset-botBot commented Aug 28, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d04beb2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblack

joshblack commented Aug 28, 2023

Copy link
Copy Markdown
Member

@radglob what does the splash zone of this change look like upstream? Are there any changes we need to make in dotcom before landing this?

@github-actions

github-actionsBot commented Aug 28, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.65 KB (-0.01% 🔽)
dist/browser.umd.js105.21 KB (-0.01% 🔽)

@radglob

Copy link
Copy Markdown
ContributorAuthor

@radglob what does the splash zone of this change look like upstream? Are there any changes we need to make in dotcom before landing this?

I don't think we have any issues upstream at this point. Armagan had looked into it a bit, and the issue in dotcom was a test that was checking for a main landmark in the react-app element, but it was removed so should no longer be an issue.

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3675 August 28, 2023 15:32 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3675 August 28, 2023 15:35 Inactive
@joshblackjoshblack mentioned this pull request Aug 28, 2023
7 tasks
@joshblack

Copy link
Copy Markdown
Member

@radglob sweet, sounds great! Opened up: https://github.com/github/github/pull/286317 for the last round of checks but otherwise LGTM! Will leave an approval on the PR once that goes green ✅

@radglob
radglob added this pull request to the merge queueAug 29, 2023
Merged via the queue into main with commit ade10e6Aug 29, 2023
@radglob
radglob deleted the remove-pagelayout-content-main-landmark branch August 29, 2023 15:53
@primerprimerBot mentioned this pull request Aug 28, 2023
@TylerJDevTylerJDev mentioned this pull request Aug 29, 2023
7 tasks
@primerprimerBot mentioned this pull request Aug 28, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@radglob@joshblack