Skip to content

Add minWidth prop to PageLayout.Pane - #3229

Merged
joshblack merged 5 commits into
mainfrom
pagelayout-min-width
May 1, 2023
Merged

Add minWidth prop to PageLayout.Pane#3229
joshblack merged 5 commits into
mainfrom
pagelayout-min-width

Conversation

@colebemis

Copy link
Copy Markdown
Contributor

Closes#3116

@changeset-bot

changeset-botBot commented Apr 26, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 86dad15

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Apr 26, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js121.18 KB (+0.01% 🔺)
dist/browser.umd.js121.48 KB (+0.01% 🔺)

@github-actions
github-actionsBot temporarily deployed to storybook-preview-3229 April 26, 2023 19:06 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3229 April 26, 2023 19:07 Inactive
@colebemis
colebemis marked this pull request as ready for review April 26, 2023 19:07
@colebemis
colebemis requested review from a team and siddharthkpApril 26, 2023 19:07
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3229 April 26, 2023 19:09 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-3229 April 26, 2023 19:15 Inactive
@colebemis
colebemis added this pull request to the merge queueApr 27, 2023
@github-merge-queue
github-merge-queueBot removed this pull request from the merge queue due to failed status checks Apr 27, 2023

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@joshblack
joshblack added this pull request to the merge queueMay 1, 2023
Merged via the queue into main with commit c0cbdd0May 1, 2023
@joshblack
joshblack deleted the pagelayout-min-width branch May 1, 2023 15:02
This was referenced May 1, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

SplitPageLayout.Pane minWidth is hardcoded

3 participants

@colebemis@siddharthkp@joshblack