Skip to content

Remove components from deprecated - #3261

Closed
broccolinisoup wants to merge 17 commits into
next-majorfrom
remove-from-deprecated
Closed

Remove components from deprecated#3261
broccolinisoup wants to merge 17 commits into
next-majorfrom
remove-from-deprecated

Conversation

@broccolinisoup

Copy link
Copy Markdown
Member

@changeset-bot

changeset-botBot commented May 8, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 51cb0df

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMajor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@broccolinisoupbroccolinisoup added the major release breaking changes label May 8, 2023
@github-actions

github-actionsBot commented May 8, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js120.38 KB (-0.72% 🔽)
dist/browser.umd.js120.61 KB (-0.76% 🔽)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 8, 2023 08:41 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 8, 2023 08:41 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 8, 2023 08:51 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 8, 2023 08:51 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 01:49 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 01:49 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 02:07 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 02:08 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 02:28 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 02:28 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 05:23 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 05:23 Inactive
@broccolinisoup
broccolinisoupforce-pushed the remove-from-deprecated branch from 9318ab6 to 5ed1f59CompareMay 9, 2023 05:32
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 05:33 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 05:39 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 05:39 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 9, 2023 06:33 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 9, 2023 06:33 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages May 10, 2023 02:37 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3261 May 10, 2023 02:37 Inactive
@broccolinisoup

Copy link
Copy Markdown
MemberAuthor

Awaiting this PR to be merged #3247 that updates the deprecated Button to the new one.

@jonrohan
jonrohan deleted the remove-from-deprecated branch May 9, 2025 18:34
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

major releasebreaking changes

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant

@broccolinisoup