Skip to content

Tooltip: Tooltip V2 as an experimental/draft with the new popover API - #3394

Merged
broccolinisoup merged 61 commits into
mainfrom
bs/experimental-tooltip-with-popover
Sep 26, 2023
Merged

Tooltip: Tooltip V2 as an experimental/draft with the new popover API#3394
broccolinisoup merged 61 commits into
mainfrom
bs/experimental-tooltip-with-popover

Conversation

@broccolinisoup

@broccolinisoupbroccolinisoup commented Jun 9, 2023

Copy link
Copy Markdown
Member

Describe your changes here.

This PR implements the new popover API (with polyfills), updates the components API that was accepted on https://github.com/github/primer/pull/2339 and addresses the accessibility remediations that were found on https://github.com/github/primer/issues/1790.

Integration PR at dotcom: https://github.com/github/github/pull/288842 (Hubbers link only)

Huge thanks to @keithamus for pairing with me on the new popover API! ✨ This is the first popover implementation on React so I would really love your thoughts and feedback 🙏🏻

TODO

  • Sort out the CSS polyfill
  • Tooltip should work within ActionMenu.Anchor when the trigger is a button
  • Update the documentation once get a 👍🏻 on the implementation
  • Add a changeset

Screenshots

Please see the storybook examples here: https://primer-dc05b46969-13348165.drafts.github.io/storybook/?path=/story/drafts-components-tooltip--default
(Currently there is no caret on the tooltip as we are in the process of proposing this as a new visual design - reference)

Merge checklist

  • Added/updated tests
  • Added/updated documentation
  • Changes are SSR compatible
  • Tested in Chrome
  • Tested in Firefox
  • Tested in Safari
  • Tested in Edge

Take a look at the What we look for in reviews section of the contributing guidelines for more information on how we review PRs.

@broccolinisoupbroccolinisoup added the skip changeset This change does not need a changelog label Jun 9, 2023
@changeset-bot

changeset-botBot commented Jun 9, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 133246d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Jun 9, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.36 KB (+3.6% 🔺)
dist/browser.umd.js104.94 KB (+3.58% 🔺)

@broccolinisoup
broccolinisoup temporarily deployed to github-pages June 9, 2023 09:35 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 June 9, 2023 09:36 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages July 20, 2023 07:19 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 July 20, 2023 07:19 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages July 24, 2023 05:26 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 July 24, 2023 05:27 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages July 24, 2023 05:43 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 July 24, 2023 05:44 Inactive
@broccolinisoup
broccolinisoup temporarily deployed to github-pages July 24, 2023 07:59 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 July 24, 2023 07:59 Inactive
@broccolinisoup
broccolinisoup marked this pull request as ready for review July 24, 2023 08:30
@broccolinisoup
broccolinisoup requested review from a team and mperrottiJuly 24, 2023 08:31
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Sep 18, 2023
@primer
primerBot temporarily deployed to github-pages September 18, 2023 01:44 Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 September 18, 2023 01:44 Inactive
@broccolinisoup
broccolinisoupforce-pushed the bs/experimental-tooltip-with-popover branch from 033c663 to 782dbdfCompareSeptember 18, 2023 03:50
@broccolinisoup
broccolinisoup temporarily deployed to github-pages September 18, 2023 03:56 — with GitHub Actions Inactive
@github-actions
github-actionsBot temporarily deployed to storybook-preview-3394 September 18, 2023 03:56 Inactive

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looking great now! Ship it!

Comment threaddocs/gatsby-config.js Outdated
@primerprimerBot mentioned this pull request Sep 26, 2023
@primerprimerBot mentioned this pull request Sep 28, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Tooltip should use Popover API to avoid z-index issues Tooltips not taking into account the margin of their anchor

3 participants

@broccolinisoup@siddharthkp@langermank