Skip to content

Replace deprecated Button in Dialog - #4049

Merged
TylerJDev merged 8 commits into
mainfrom
tylerjdev/replace-deprecated-buttons-dialog
Dec 19, 2023
Merged

Replace deprecated Button in Dialog#4049
TylerJDev merged 8 commits into
mainfrom
tylerjdev/replace-deprecated-buttons-dialog

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Dec 11, 2023

Copy link
Copy Markdown
Member

Closeshttps://github.com/github/primer/issues/2470, #3062

Changelog

Changed

Replaces deprecated Button component in Dialog.tsx with newest version.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

For a successful rollout in github, we'll need to adjust some items that rely on the deprecated version of Button. The changes needed can be viewed in the integration test PR.

Testing & Reviewing

Ensure that a focus indicator is present when focused on any of the dialog's footer buttons.

Dialog default Storybook example.

Merge checklist

@changeset-bot

changeset-botBot commented Dec 11, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 61df519

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Dec 11, 2023

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
dist/browser.esm.js104.71 KB (-0.8% 🔽)
dist/browser.umd.js105.27 KB (-0.82% 🔽)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4049 December 11, 2023 21:24 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review December 12, 2023 21:35
@TylerJDev
TylerJDev requested review from a team and mperrottiDecember 12, 2023 21:35
Comment thread.changeset/eight-pillows-pump.md Outdated
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4049 December 14, 2023 15:03 Inactive
@TylerJDevTylerJDev added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 14, 2023
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Dec 14, 2023

@joshblackjoshblack left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

These changes look great to me and the upstream PR makes it seem like we only need to update tests 💯 I'll also request @broccolinisoup who I think may also have some helpful context for migrating this work if I remember right (I think she had helped an outside contributor trying to do this in the past and I wasn't sure what the hiccups ended up being there)

@broccolinisoup

Copy link
Copy Markdown
Member

These changes look great to me and the upstream PR makes it seem like we only need to update tests 💯 I'll also request @broccolinisoup who I think may also have some helpful context for migrating this work if I remember right (I think she had helped an outside contributor trying to do this in the past and I wasn't sure what the hiccups ended up being there)

Thanks @joshblack for the ping, I appreciate it! We chatted about it in our call with Tyler and the reason that we reverted this PR is addressed in the integration PR 🤗 And the solution is approved by the code owners. Seems like we are good to go!

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great! Thank you so much for fixing that and providing the integration fixes as well ✨

@TylerJDev
TylerJDev added this pull request to the merge queueDec 19, 2023
Merged via the queue into main with commit f0d38bcDec 19, 2023
@TylerJDev
TylerJDev deleted the tylerjdev/replace-deprecated-buttons-dialog branch December 19, 2023 14:33
@primerprimerBot mentioned this pull request Dec 19, 2023
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@TylerJDev@broccolinisoup@joshblack