Skip to content

Address a few v8 color bugs - #4278

Merged
langermank merged 11 commits into
mainfrom
color-token-fixes
Feb 20, 2024
Merged

Address a few v8 color bugs#4278
langermank merged 11 commits into
mainfrom
color-token-fixes

Conversation

@langermank

Copy link
Copy Markdown
Contributor

I ran an integration test with v8 enabled in Storybook and found a few color bugs.

  • Fix Counter bg color and border
  • Various control disabled states

@langermank
langermank requested review from a team and siddharthkpFebruary 15, 2024 22:55
@changeset-bot

changeset-botBot commented Feb 15, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9968ed8

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Feb 15, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.47 KB (+0.09% 🔺)
packages/react/dist/browser.umd.js114.14 KB (+0.11% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4278 February 15, 2024 22:59 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 15, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4278 February 16, 2024 00:09 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4278 February 16, 2024 00:41 Inactive
@langermanklangermank added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 16, 2024
@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4278 February 16, 2024 01:30 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4278 February 20, 2024 20:15 Inactive
@langermank
langermank added this pull request to the merge queueFeb 20, 2024
Merged via the queue into main with commit 9ec7262Feb 20, 2024
@langermank
langermank deleted the color-token-fixes branch February 20, 2024 20:55
@primerprimerBot mentioned this pull request Feb 20, 2024
TylerJDev pushed a commit that referenced this pull request Feb 20, 2024
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
TylerJDev pushed a commit that referenced this pull request Mar 14, 2024
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
TylerJDev pushed a commit that referenced this pull request Mar 27, 2024
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
TylerJDev pushed a commit that referenced this pull request Mar 27, 2024
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
github-merge-queueBot pushed a commit that referenced this pull request Jun 18, 2024
* Add fixes for ActionList
* More type fixes
* temp type fixes
* Add condition for inactive items
* add yet another condition
* chore(deps): bump changesets/action from 1.4.5 to 1.4.6 (#4282)
Bumps [changesets/action](https://github.com/changesets/action) from 1.4.5 to 1.4.6.
- [Release notes](https://github.com/changesets/action/releases)
- [Changelog](https://github.com/changesets/action/blob/main/CHANGELOG.md)
- [Commits](changesets/action@f13b1ba...e2f8e96)
---
updated-dependencies:
- dependency-name: changesets/action
dependency-type: direct:production
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* test(e2e): add e2e test for SelectPanel2 default story (#4279)
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Address a few v8 color bugs (#4278)
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* chore(deps-dev): bump ip from 2.0.0 to 2.0.1 (#4291)
Bumps [ip](https://github.com/indutny/node-ip) from 2.0.0 to 2.0.1.
- [Commits](indutny/node-ip@v2.0.0...v2.0.1)
---
updated-dependencies:
- dependency-name: ip
dependency-type: indirect
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* Change how `ref` is handled
* Change types
* Update storybook example types
* Update types on component
* Add another type
* Update type in `SegmentedControl`
* Add back `li`-only type
* Add onto `onSelect` type
* Update more types
* Type fixes for `LinkItem`
* Changes from feedback
* Change types
* Replace `role="list"` with context
* Add feature flag to `ActionList.Item
* Add back forwardedRef in cases where valid role is true
* Update FF name
* Add lint disable
* Update FF name
* Add changeset
* Remove `list` condition
* Rename FF
* Address feedback
* Add feature flag story
* Add new test
* Add feature flag to all stories
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
github-merge-queueBot pushed a commit that referenced this pull request Jun 21, 2024
* Add fixes for ActionList
* More type fixes
* temp type fixes
* Add condition for inactive items
* add yet another condition
* chore(deps): bump changesets/action from 1.4.5 to 1.4.6 (#4282)
Bumps [changesets/action](https://github.com/changesets/action) from 1.4.5 to 1.4.6.
- [Release notes](https://github.com/changesets/action/releases)
- [Changelog](https://github.com/changesets/action/blob/main/CHANGELOG.md)
- [Commits](changesets/action@f13b1ba...e2f8e96)
---
updated-dependencies:
- dependency-name: changesets/action
dependency-type: direct:production
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* test(e2e): add e2e test for SelectPanel2 default story (#4279)
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Address a few v8 color bugs (#4278)
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* chore(deps-dev): bump ip from 2.0.0 to 2.0.1 (#4291)
Bumps [ip](https://github.com/indutny/node-ip) from 2.0.0 to 2.0.1.
- [Commits](indutny/node-ip@v2.0.0...v2.0.1)
---
updated-dependencies:
- dependency-name: ip
dependency-type: indirect
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* Change how `ref` is handled
* Change types
* Update storybook example types
* Update types on component
* Add another type
* Update type in `SegmentedControl`
* Add back `li`-only type
* Add onto `onSelect` type
* Update more types
* Type fixes for `LinkItem`
* Changes from feedback
* Change types
* Replace `role="list"` with context
* Add feature flag to `ActionList.Item
* Add back forwardedRef in cases where valid role is true
* Update FF name
* Add lint disable
* Update FF name
* Add changeset
* Remove `list` condition
* Rename FF
* Address feedback
* Add feature flag story
* Add new test
* Add export
* Updated forwardRef to be polymorphic
* Spike: Add trailingAction to ActionList.Item
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* Updated forwardRef to be polymorphic
* TrailingAction style adjustments
* test(vrt): update snapshots
* Remove unused story
* Limit TrailingAction props
* Fix missing href for as link
* Add Inactive example
* Add styles for `showOnHover`
* Fixed href overload issue
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* Made sure that TrailingAction doesn't show when inactive state
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* test(vrt): update snapshots
* Add fixes for button styles
* Fix for flex styles
* Update storybookd descriptions
* Add block description example
* Add tests
* Remove FF from tests
* update e2e stories
* spike on allowing text (#4659)
* spike on allowing text
* showOnHover support
* Add overflow example
* Add visual tests
* Update from display to visibility
* Create calm-crabs-raise.md
* Disallow usage in `ActionMenu`
* Update ActionList.test.tsx
* Remove hover/focus styles
* test(vrt): update snapshots
* Apply suggestions from code review
* Apply suggestions from code review
* Fix merge conflict
* Remove Hover examples
* Remove showOnHover code
* Add FF const back
* Remove some usage of `hoverStyles`
* test(vrt): update snapshots
* Add props to TrailingAction
* update types
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* update docs
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: Tyler Jones <tylerjdev@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: khiga8 <khiga8@users.noreply.github.com>
Co-authored-by: TylerJDev <TylerJDev@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@langermank@joshblack