Skip to content

test(e2e): add e2e test for SelectPanel2 default story - #4279

Merged
joshblack merged 4 commits into
mainfrom
3025-add-e2e-tests-select-panel
Feb 20, 2024
Merged

test(e2e): add e2e test for SelectPanel2 default story#4279
joshblack merged 4 commits into
mainfrom
3025-add-e2e-tests-select-panel

Conversation

@joshblack

Copy link
Copy Markdown
Member

Context https://github.com/github/primer/issues/3025

This PR adds the first e2e test for SelectPanel2 for its default story. It also addresses the axe violation that shows up for the role="listbox" by labeling it by the title id for the dialog

Changelog

New

  • Add e2e test for SelectPanel2

Changed

  • Add listLabelledBy as context value for ActionList

Removed

Rollout strategy

  • Patch release

Testing & Reviewing

  • Does the change for the aria-labelledby make sense? will this always be available or is this something teams should provide?

@joshblack
joshblack requested review from a team and siddharthkpFebruary 16, 2024 20:20
@changeset-bot

changeset-botBot commented Feb 16, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9bec151

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@joshblackjoshblack added the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 16, 2024
@joshblackjoshblack changed the title 3025 add e2e tests select paneltest(e2e): add e2e test for SelectPanel2 default storyFeb 16, 2024
@github-actions

github-actionsBot commented Feb 16, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js113.37 KB (0%)
packages/react/dist/browser.umd.js114.03 KB (0%)

@github-actionsgithub-actionsBot removed the update snapshots 🤖 Command that updates VRT snapshots on the pull request label Feb 16, 2024
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4279 February 16, 2024 20:32 Inactive
selectionAttribute: 'aria-selected',
selectionVariant: selectionVariant === 'instant' ? 'single' : selectionVariant,
afterSelect: internalAfterSelect,
listLabelledBy: `${panelId}--title`,

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

very nice!

@siddharthkpsiddharthkp left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

These look great!

@joshblack
joshblack added this pull request to the merge queueFeb 20, 2024
Merged via the queue into main with commit a68e2e2Feb 20, 2024
@joshblack
joshblack deleted the 3025-add-e2e-tests-select-panel branch February 20, 2024 19:11
@primerprimerBot mentioned this pull request Feb 20, 2024
TylerJDev pushed a commit that referenced this pull request Feb 20, 2024
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
github-merge-queueBot pushed a commit that referenced this pull request Jun 18, 2024
* Add fixes for ActionList
* More type fixes
* temp type fixes
* Add condition for inactive items
* add yet another condition
* chore(deps): bump changesets/action from 1.4.5 to 1.4.6 (#4282)
Bumps [changesets/action](https://github.com/changesets/action) from 1.4.5 to 1.4.6.
- [Release notes](https://github.com/changesets/action/releases)
- [Changelog](https://github.com/changesets/action/blob/main/CHANGELOG.md)
- [Commits](changesets/action@f13b1ba...e2f8e96)
---
updated-dependencies:
- dependency-name: changesets/action
dependency-type: direct:production
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* test(e2e): add e2e test for SelectPanel2 default story (#4279)
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Address a few v8 color bugs (#4278)
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* chore(deps-dev): bump ip from 2.0.0 to 2.0.1 (#4291)
Bumps [ip](https://github.com/indutny/node-ip) from 2.0.0 to 2.0.1.
- [Commits](indutny/node-ip@v2.0.0...v2.0.1)
---
updated-dependencies:
- dependency-name: ip
dependency-type: indirect
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* Change how `ref` is handled
* Change types
* Update storybook example types
* Update types on component
* Add another type
* Update type in `SegmentedControl`
* Add back `li`-only type
* Add onto `onSelect` type
* Update more types
* Type fixes for `LinkItem`
* Changes from feedback
* Change types
* Replace `role="list"` with context
* Add feature flag to `ActionList.Item
* Add back forwardedRef in cases where valid role is true
* Update FF name
* Add lint disable
* Update FF name
* Add changeset
* Remove `list` condition
* Rename FF
* Address feedback
* Add feature flag story
* Add new test
* Add feature flag to all stories
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
github-merge-queueBot pushed a commit that referenced this pull request Jun 21, 2024
* Add fixes for ActionList
* More type fixes
* temp type fixes
* Add condition for inactive items
* add yet another condition
* chore(deps): bump changesets/action from 1.4.5 to 1.4.6 (#4282)
Bumps [changesets/action](https://github.com/changesets/action) from 1.4.5 to 1.4.6.
- [Release notes](https://github.com/changesets/action/releases)
- [Changelog](https://github.com/changesets/action/blob/main/CHANGELOG.md)
- [Commits](changesets/action@f13b1ba...e2f8e96)
---
updated-dependencies:
- dependency-name: changesets/action
dependency-type: direct:production
update-type: version-update:semver-patch
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* test(e2e): add e2e test for SelectPanel2 default story (#4279)
* fix(SelectPanel2): add aria-labelledby to listbox
* test(e2e): add e2e test for SelectPanel2 default story
* chore: add changeset
* test(vrt): update snapshots
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* Address a few v8 color bugs (#4278)
* small bug fixes with v8
* Create khaki-schools-lay.md
* test(vrt): update snapshots
* snippy snaps
* test(vrt): update snapshots
* test: update snapshots
* test: update snapshots
* try commenting flakey tests
* test: comment out flakey snapshot test
---------
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* chore(deps-dev): bump ip from 2.0.0 to 2.0.1 (#4291)
Bumps [ip](https://github.com/indutny/node-ip) from 2.0.0 to 2.0.1.
- [Commits](indutny/node-ip@v2.0.0...v2.0.1)
---
updated-dependencies:
- dependency-name: ip
dependency-type: indirect
...
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
* Change how `ref` is handled
* Change types
* Update storybook example types
* Update types on component
* Add another type
* Update type in `SegmentedControl`
* Add back `li`-only type
* Add onto `onSelect` type
* Update more types
* Type fixes for `LinkItem`
* Changes from feedback
* Change types
* Replace `role="list"` with context
* Add feature flag to `ActionList.Item
* Add back forwardedRef in cases where valid role is true
* Update FF name
* Add lint disable
* Update FF name
* Add changeset
* Remove `list` condition
* Rename FF
* Address feedback
* Add feature flag story
* Add new test
* Add export
* Updated forwardRef to be polymorphic
* Spike: Add trailingAction to ActionList.Item
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* Updated forwardRef to be polymorphic
* TrailingAction style adjustments
* test(vrt): update snapshots
* Remove unused story
* Limit TrailingAction props
* Fix missing href for as link
* Add Inactive example
* Add styles for `showOnHover`
* Fixed href overload issue
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* Made sure that TrailingAction doesn't show when inactive state
Co-authored-by: Tyler Jones <tylerjdev@github.com>
* test(vrt): update snapshots
* Add fixes for button styles
* Fix for flex styles
* Update storybookd descriptions
* Add block description example
* Add tests
* Remove FF from tests
* update e2e stories
* spike on allowing text (#4659)
* spike on allowing text
* showOnHover support
* Add overflow example
* Add visual tests
* Update from display to visibility
* Create calm-crabs-raise.md
* Disallow usage in `ActionMenu`
* Update ActionList.test.tsx
* Remove hover/focus styles
* test(vrt): update snapshots
* Apply suggestions from code review
* Apply suggestions from code review
* Fix merge conflict
* Remove Hover examples
* Remove showOnHover code
* Add FF const back
* Remove some usage of `hoverStyles`
* test(vrt): update snapshots
* Add props to TrailingAction
* update types
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
* update docs
---------
Signed-off-by: dependabot[bot] <support@github.com>
Co-authored-by: Tyler Jones <tylerjdev@github.com>
Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com>
Co-authored-by: Josh Black <joshblack@github.com>
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Katie Langerman <18661030+langermank@users.noreply.github.com>
Co-authored-by: langermank <langermank@users.noreply.github.com>
Co-authored-by: khiga8 <khiga8@users.noreply.github.com>
Co-authored-by: TylerJDev <TylerJDev@users.noreply.github.com>
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@joshblack@siddharthkp