Skip to content

Adds aria-expanded="false" back to AnchoredOverlay - #4394

Merged
TylerJDev merged 2 commits into
mainfrom
tylerjdev/add-aria-expanded-state
Mar 19, 2024
Merged

Adds aria-expanded="false" back to AnchoredOverlay#4394
TylerJDev merged 2 commits into
mainfrom
tylerjdev/add-aria-expanded-state

Conversation

@TylerJDev

@TylerJDevTylerJDev commented Mar 14, 2024

Copy link
Copy Markdown
Member

Will resolve: https://github.com/github/accessibility-audits/issues/6932

We have an open accessibility issue related to how we handle aria-expanded within AnchoredOverlay. We did not apply aria-expanded when a menu was open since it was redundant, and it was very unlikely that there'd be navigation outside of the menu before it was closed. We've received feedback that this may cause an unintended effect for users using specific devices/AT that makes it difficult to leave the menu.

Changelog

Changed

  • Adds full aria-expanded state to AnchoredOverlay.

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Testing & Reviewing

Merge checklist

@changeset-bot

changeset-botBot commented Mar 14, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7cc9d42

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actionsBot commented Mar 14, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js87.38 KB (-0.16% 🔽)
packages/react/dist/browser.umd.js87.75 KB (+0.03% 🔺)

@github-actions
github-actionsBottemporarily deployed to storybook-preview-4394 March 14, 2024 19:21 Inactive
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4394 March 14, 2024 19:34 Inactive
@TylerJDev
TylerJDev marked this pull request as ready for review March 14, 2024 20:01
@TylerJDev
TylerJDev requested a review from a team as a code ownerMarch 14, 2024 20:01
@TylerJDev
TylerJDev added this pull request to the merge queueMar 19, 2024
Merged via the queue into main with commit cc9b98aMar 19, 2024
@TylerJDev
TylerJDev deleted the tylerjdev/add-aria-expanded-state branch March 19, 2024 13:38
@primerprimerBot mentioned this pull request Mar 19, 2024
@siddharthkpsiddharthkp mentioned this pull request Apr 2, 2024
4 tasks
siddharthkp added a commit that referenced this pull request Apr 2, 2024
github-merge-queueBot pushed a commit that referenced this pull request Apr 2, 2024
lukasoppermann pushed a commit that referenced this pull request Apr 16, 2024
* Adds `aria-expanded="false"` back to `AnchoredOverlay`
* Add changeset
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants

@TylerJDev@joshblack