Skip to content

export FilteredActionList - #4816

Merged
siddharthkp merged 5 commits into
mainfrom
export-filtered-actionlist
Aug 8, 2024
Merged

export FilteredActionList#4816
siddharthkp merged 5 commits into
mainfrom
export-filtered-actionlist

Conversation

@siddharthkp

@siddharthkpsiddharthkp commented Aug 6, 2024

Copy link
Copy Markdown
Member

Closes#4840

Changelog

While work on SelectPanel is on-going, there is a gap between the design of SelectPanel and the implementation in PRC.

In order to fill the gap with a temporary solution, developers can opt to build a custom implementation. Exporting FilteredActionList, which lies under SelectPanel, would help.

More context on Slack: https://github.slack.com/archives/CSGAVNZ19/p1722543563922769?thread_ts=1722472095.221459&cid=CSGAVNZ19

Rollout strategy

  • Patch release
  • Minor release
  • Major release; if selected, include a written rollout or migration plan
  • None; if selected, include a brief description as to why

Merge checklist

@changeset-bot

changeset-botBot commented Aug 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 839501c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactPatch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@siddharthkpsiddharthkp self-assigned this Aug 6, 2024
@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 6, 2024
@github-actions

github-actionsBot commented Aug 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96 KB (0%)
packages/react/dist/browser.umd.js96.31 KB (0%)

Comment threadpackages/react/src/index.ts Outdated
export type {SxProp} from './sx'

export {FilteredActionList} from './FilteredActionList'
export type {FilteredActionListProps} from './FilteredActionList'

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Question for reviewer:

Should we export this from experimental instead? It isn't really experimental because it's been what powers SelectPanel from the start.

But if we want to treat this as a temporary solution, then it feels okay

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I'm down with experimental if it means we won't have to go through a deprecated flow 😂 But if we still have to go through that, then stable is fine 👍

(Separate from this PR) this could be a good example for spots where an unstable_* convention could help out without having to couple things to entry points

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

My only comment is that it would be great to treat it as an experimental/unstable component so that we don't have to worry about semver in case if we need to make any changes for the SelectPanel remediation.

@siddharthkpsiddharthkpAug 8, 2024

Copy link
Copy Markdown
MemberAuthor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Valid, moved it to experimental

@siddharthkp
siddharthkp marked this pull request as ready for review August 6, 2024 10:37
@siddharthkp
siddharthkp requested a review from a team as a code ownerAugust 6, 2024 10:37

@broccolinisoupbroccolinisoup left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Left a comment! Let me know your thoughts!

@siddharthkp
siddharthkp added this pull request to the merge queueAug 8, 2024
Merged via the queue into main with commit f9bc73cAug 8, 2024
@siddharthkp
siddharthkp deleted the export-filtered-actionlist branch August 8, 2024 20:23
@primerprimerBot mentioned this pull request Aug 8, 2024
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Need access to FilteredActionList to create a stopgap modal SelectPanel

3 participants

@siddharthkp@broccolinisoup@joshblack