Skip to content

refactor(Heading): update component to CSS Modules - #4819

Merged
jonrohan merged 12 commits into
mainfrom
3716-refactor-heading-to-css-modules
Aug 14, 2024
Merged

refactor(Heading): update component to CSS Modules#4819
jonrohan merged 12 commits into
mainfrom
3716-refactor-heading-to-css-modules

Conversation

@joshblack

@joshblackjoshblack commented Aug 6, 2024

Copy link
Copy Markdown
Member

Part of https://github.com/github/primer/issues/3716

Changelog

New

Changed

  • Refactor Heading to use CSS Modules

Removed

Rollout strategy

  • Minor release

@joshblack
joshblack requested a review from a team as a code ownerAugust 6, 2024 20:52
@changeset-bot

changeset-botBot commented Aug 6, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 61e16dd

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
NameType
@primer/reactMinor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actionsgithub-actionsBot added the staff Author is a staff member label Aug 6, 2024
@github-actions

github-actionsBot commented Aug 6, 2024

Copy link
Copy Markdown
Contributor

size-limit report 📦

PathSize
packages/react/dist/browser.esm.js96.16 KB (+0.1% 🔺)
packages/react/dist/browser.umd.js96.42 KB (-0.02% 🔽)

@joshblack
joshblack requested a review from a team as a code ownerAugust 7, 2024 20:21
Comment threadpackages/react/src/Heading/Heading.tsx Outdated
@joshblack

Copy link
Copy Markdown
MemberAuthor

Looks like examples failure is the same thing you run into @jonrohan over in #4828 👀 Will wait for that to merge 👍

@jonrohan
jonrohan enabled auto-merge August 14, 2024 22:04
@github-actions
github-actionsBottemporarily deployed to storybook-preview-4819 August 14, 2024 22:26 Inactive
@jonrohan
jonrohan added this pull request to the merge queueAug 14, 2024
Merged via the queue into main with commit 0112347Aug 14, 2024
@jonrohan
jonrohan deleted the 3716-refactor-heading-to-css-modules branch August 14, 2024 22:47
@primerprimerBot mentioned this pull request Aug 14, 2024
jonrohan added a commit that referenced this pull request Aug 15, 2024
* refactor(Heading): update component to CSS Modules
* test(e2e): add feature flags for Heading test
* chore: fix stylelint errors
* test(e2e): update snapshots for Heading
* refactor(Heading): incorporate variant features into CSS Modules
* chore: add changeset
* test(e2e): remove theme testing
* Fix props here
* Remove examples/nextjs
* Clean up css
---------
Co-authored-by: Josh Black <joshblack@users.noreply.github.com>
Co-authored-by: Jon Rohan <yes@jonrohan.codes>
@primerprimerBot mentioned this pull request Oct 18, 2024
Sign up for freeto join this conversation on GitHub. Already have an account? Sign in to comment

Labels

staffAuthor is a staff member

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants

@joshblack@jonrohan@langermank